Conversation
kaovilai
force-pushed
the
just_registryCanUseOtherSecretNameFor_credentialsFile
branch
from
April 29, 2022 01:59
204f2d9 to
a05909a
Compare
kaovilai
force-pushed
the
just_registryCanUseOtherSecretNameFor_credentialsFile
branch
from
April 29, 2022 05:15
a05909a to
5a2bf0b
Compare
Member
Author
|
/retest |
3 similar comments
Member
Author
|
/retest |
Member
Author
|
/retest |
Member
Author
|
/retest |
Member
Author
|
/retest |
|
@kaovilai: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
Member
Author
|
5 retest later. Beautiful |
kaovilai
requested review from
deepakraj1997 and
shubham-pampattiwar
and removed request for
JaydipGabani and
shawn-hurley
April 29, 2022 21:24
Member
Author
|
Closing in favor of #652 |
12 tasks
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
OADP-477