[release-4.14] OCPBUGS-23558: Let router use svc ips 4.14 - #3221
Conversation
The kas SVC port is an impl detail. Before this change their value was wrongly conflated with the spec.metworking.apiServer.port input. If we ever require the dedicated KAS LB port to be a choice, that would be input in the LB publishing strategy.
|
@enxebre: This pull request references Jira Issue OCPBUGS-23558, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: enxebre The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/override "Red Hat Trusted App Test / mce-24-snc9n / tektok-runner-enterprise-contract-mce-2.4" |
|
@enxebre: Overrode contexts on behalf of enxebre: Red Hat Trusted App Test / mce-24-snc9n / tektok-runner-enterprise-contract-mce-2.4 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
openshift/release#45885 |
|
/test e2e-aws |
|
/lgtm |
|
/jira refresh |
|
@enxebre: This pull request references Jira Issue OCPBUGS-23558, which is invalid:
Comment DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/jira refresh |
|
@enxebre: This pull request references Jira Issue OCPBUGS-23558, which is valid. The bug has been moved to the POST state. 6 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (jiezhao@redhat.com), skipping review request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/override "Red Hat Trusted App Pipeline / hypershift-release-414-on-pull-request" |
|
@enxebre: Overrode contexts on behalf of enxebre: Red Hat Trusted App Pipeline / hypershift-release-414-on-pull-request DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
All test passed though prow complains pod go deleted unexpectedly https://prow.ci.openshift.org/view/gs/origin-ci-test/pr-logs/pull/openshift_hypershift/3221/pull-ci-openshift-hypershift-release-4.14-e2e-aws/1727597362857119744 /test e2e-aws |
|
@enxebre: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
@enxebre: Jira Issue OCPBUGS-23558: All pull requests linked via external trackers have merged: Jira Issue OCPBUGS-23558 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
Fix included in accepted release 4.14.0-0.nightly-2023-11-23-154014 |
What this PR does / why we need it:
Cherry-picks #3218
It also brings
#3185
#3186
To accommodate conflicts and reduce drift.
Which issue(s) this PR fixes (optional, use
fixes #<issue_number>(, fixes #<issue_number>, ...)format, where issue_number might be a GitHub issue, or a Jira story:Fixes #
Checklist