Skip to content

hypershift/install: configure a very short cert rotation scale - #45885

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
stevekuznetsov:skuznets/hypershift-cert-rotation-scale
Nov 21, 2023
Merged

hypershift/install: configure a very short cert rotation scale#45885
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
stevekuznetsov:skuznets/hypershift-cert-rotation-scale

Conversation

@stevekuznetsov

Copy link
Copy Markdown
Contributor

We want to ensure that cert rotation is happening during normal operations in CI, so let's set a very short scale (about 1/60th of normal).

We want to ensure that cert rotation is happening during normal
operations in CI, so let's set a very short scale (about 1/60th of
normal).

Signed-off-by: Steve Kuznetsov <skuznets@redhat.com>
@openshift-ci
openshift-ci Bot requested review from imain and sjenning November 20, 2023 22:07
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@stevekuznetsov: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-kubevirt-csi-driver-master-e2e-kubevirt-aws-csi openshift/kubevirt-csi-driver presubmit Registry content changed
pull-ci-openshift-kubevirt-csi-driver-release-4.16-e2e-kubevirt-aws-csi openshift/kubevirt-csi-driver presubmit Registry content changed
pull-ci-openshift-kubevirt-csi-driver-release-4.15-e2e-kubevirt-aws-csi openshift/kubevirt-csi-driver presubmit Registry content changed
pull-ci-openshift-kubevirt-csi-driver-release-4.14-e2e-kubevirt-aws-csi openshift/kubevirt-csi-driver presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.13-e2e-hypershift-ovn openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.12-e2e-hypershift-ovn openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.11-e2e-hypershift openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-master-e2e-hypershift-ovn openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.16-e2e-hypershift-ovn openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.15-e2e-hypershift-ovn openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.14-e2e-hypershift-ovn openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-master-e2e-aws-hypershift-ovn-kubevirt openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.16-e2e-aws-hypershift-ovn-kubevirt openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.15-e2e-aws-hypershift-ovn-kubevirt openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-hypershift-ovn-kubevirt openshift/cluster-network-operator presubmit Registry content changed
pull-ci-openshift-csi-operator-master-hypershift-aws-e2e-external openshift/csi-operator presubmit Registry content changed
pull-ci-openshift-csi-operator-release-4.16-hypershift-aws-e2e-external openshift/csi-operator presubmit Registry content changed
pull-ci-openshift-csi-operator-release-4.15-hypershift-aws-e2e-external openshift/csi-operator presubmit Registry content changed
pull-ci-openshift-cluster-node-tuning-operator-master-e2e-hypershift openshift/cluster-node-tuning-operator presubmit Registry content changed
pull-ci-openshift-cluster-node-tuning-operator-release-4.16-e2e-hypershift openshift/cluster-node-tuning-operator presubmit Registry content changed
pull-ci-openshift-cluster-node-tuning-operator-release-4.15-e2e-hypershift openshift/cluster-node-tuning-operator presubmit Registry content changed
pull-ci-openshift-cluster-node-tuning-operator-release-4.14-e2e-hypershift openshift/cluster-node-tuning-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.13-e2e-hypershift-ovn-conformance openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-master-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.16-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.15-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.14-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cluster-storage-operator-release-4.12-hypershift-aws-e2e-external openshift/cluster-storage-operator presubmit Registry content changed
pull-ci-openshift-cloud-provider-kubevirt-main-e2e-kubevirt-aws-ovn openshift/cloud-provider-kubevirt presubmit Registry content changed
pull-ci-openshift-cloud-provider-kubevirt-release-4.16-e2e-kubevirt-aws-ovn openshift/cloud-provider-kubevirt presubmit Registry content changed
pull-ci-openshift-cloud-provider-kubevirt-release-4.15-e2e-kubevirt-aws-ovn openshift/cloud-provider-kubevirt presubmit Registry content changed
pull-ci-openshift-cloud-provider-kubevirt-release-4.14-e2e-kubevirt-aws-ovn openshift/cloud-provider-kubevirt presubmit Registry content changed
pull-ci-openshift-hypershift-main-e2e-conformance openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.16-e2e-conformance openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.15-e2e-conformance openshift/hypershift presubmit Registry content changed

A total of 194 jobs have been affected by this change. The above listing is non-exhaustive and limited to 35 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 10 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 20 rehearsals
Comment: /pj-rehearse max to run up to 35 rehearsals
Comment: /pj-rehearse auto-ack to run up to 10 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse abort to abort all active rehearsals

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@stevekuznetsov

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Nov 20, 2023
@sjenning

Copy link
Copy Markdown
Contributor

/approve
/lgtm
/hold for openshift/hypershift#3208

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Nov 21, 2023
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Nov 21, 2023
@openshift-ci

openshift-ci Bot commented Nov 21, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: sjenning, stevekuznetsov

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 21, 2023
@stevekuznetsov

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Nov 21, 2023
@openshift-ci

openshift-ci Bot commented Nov 21, 2023

Copy link
Copy Markdown
Contributor

@stevekuznetsov: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit bbb0e50 into openshift:master Nov 21, 2023
yxun pushed a commit to yxun/release that referenced this pull request Nov 21, 2023
…hift#45885)

We want to ensure that cert rotation is happening during normal
operations in CI, so let's set a very short scale (about 1/60th of
normal).

Signed-off-by: Steve Kuznetsov <skuznets@redhat.com>
@enxebre

enxebre commented Nov 22, 2023

Copy link
Copy Markdown
Member

@stevekuznetsov @sjenning This seems to have broken release-4.14. CI https://prow.ci.openshift.org/view/gs/origin-ci-test/pr-logs/pull/openshift_hypershift/3221/pull-ci-openshift-hypershift-release-4.14-e2e-aws/1727278742486650880

Error: unknown flag: --cert-rotation-scale
unknown flag: --cert-rotation-scale

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants