Skip to content

HOSTEDCP-1285: Consume kas Pod port by name - #3185

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
enxebre:kas-port-cleanup
Nov 13, 2023
Merged

HOSTEDCP-1285: Consume kas Pod port by name#3185
openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
enxebre:kas-port-cleanup

Conversation

@enxebre

@enxebre enxebre commented Nov 13, 2023

Copy link
Copy Markdown
Member

What this PR does / why we need it:

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

@enxebre

enxebre commented Nov 13, 2023

Copy link
Copy Markdown
Member Author

/test unit

@openshift-ci openshift-ci Bot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. do-not-merge/needs-area labels Nov 13, 2023
@openshift-ci

openshift-ci Bot commented Nov 13, 2023

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci openshift-ci Bot added area/control-plane-operator Indicates the PR includes changes for the control plane operator - in an OCP release approved Indicates a PR has been approved by an approver from all required OWNERS files. and removed do-not-merge/needs-area labels Nov 13, 2023
@enxebre enxebre changed the title Consume kas port by name HOSTEDCP-1285: Consume kas port by name Nov 13, 2023
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Nov 13, 2023
@openshift-ci-robot

openshift-ci-robot commented Nov 13, 2023

Copy link
Copy Markdown

@enxebre: This pull request references HOSTEDCP-1285 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.15.0" version, but no target version was set.

Details

In response to this:

What this PR does / why we need it:

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@enxebre
enxebre marked this pull request as ready for review November 13, 2023 10:16
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Nov 13, 2023
@openshift-ci
openshift-ci Bot requested review from hasueki and sjenning November 13, 2023 10:16
@Patryk-Stefanski

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Nov 13, 2023
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Nov 13, 2023
@enxebre enxebre changed the title HOSTEDCP-1285: Consume kas port by name HOSTEDCP-1285: Consume kas Pod port by name Nov 13, 2023
@enxebre enxebre mentioned this pull request Nov 13, 2023
4 tasks

@bryan-cox bryan-cox left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Nov 13, 2023
@openshift-ci

openshift-ci Bot commented Nov 13, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bryan-cox, enxebre

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 68990cd and 2 for PR HEAD 45b6e92 in total

@enxebre

enxebre commented Nov 13, 2023

Copy link
Copy Markdown
Member Author

/test e2e-kubevirt-aws-ovn

@openshift-ci

openshift-ci Bot commented Nov 13, 2023

Copy link
Copy Markdown
Contributor

@enxebre: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit e9e7b72 into openshift:main Nov 13, 2023
@openshift-bot

Copy link
Copy Markdown

[ART PR BUILD NOTIFIER]

This PR has been included in build ose-hypershift-container-v4.15.0-202311132249.p0.ge9e7b72.assembly.stream for distgit hypershift.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/control-plane-operator Indicates the PR includes changes for the control plane operator - in an OCP release jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants