OCPBUGS-23466: Let router use svc ips - #3218
Conversation
|
@enxebre: This pull request references Jira Issue OCPBUGS-23466, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: enxebre The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/jira refresh |
|
@enxebre: This pull request references Jira Issue OCPBUGS-23466, which is valid. The bug has been moved to the POST state. 3 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (jiezhao@redhat.com), skipping review request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
✅ Deploy Preview for hypershift-docs ready!
To edit notification comments on pull requests, go to your Netlify site configuration. |
e4efc2f to
1b2fa99
Compare
csrwng
left a comment
There was a problem hiding this comment.
Thanks, some comments, but generally lgtm
| mode http | ||
| monitor-uri /haproxy_ready | ||
|
|
||
| resolvers coredns |
| {{range .Backends }} | ||
| backend {{ .Name }} | ||
| server {{ .Name }} {{ .DestinationService }}.{{ $ns }}.svc.cluster.local:{{.DestinationPort}} check resolvers coredns init-addr last,libc,none | ||
| server {{ .Name }} {{ .DestinationServiceIP }}:{{.DestinationPort}} check resolvers coredns init-addr last,libc,none |
There was a problem hiding this comment.
s/resolvers coredns init-addr last,libc,none//
| {{- if .HasKubeAPI }} | ||
| backend kube_api | ||
| server kube_api kube-apiserver.{{ $ns }}.svc.cluster.local:{{ .KASSVCPort }} check resolvers coredns init-addr last,libc,none | ||
| server kube_api {{ .KASDestinationServiceIP }}:{{ .KASSVCPort }} check resolvers coredns init-addr last,libc,none |
There was a problem hiding this comment.
s/resolvers coredns init-addr last,libc,none//
| func (r byRouteName) Less(i, j int) bool { return r[i].Name < r[j].Name } | ||
|
|
||
| func generateRouterConfig(namespace string, routeList *routev1.RouteList, nameServerIP string) (string, error) { | ||
| func generateRouterConfig(routeList *routev1.RouteList, nameServerIP string, svcsNameToIP map[string]string) (string, error) { |
There was a problem hiding this comment.
nameServerIP is no longer needed
1b2fa99 to
1fae69c
Compare
1fae69c to
c185421
Compare
|
/lgtm |
|
@enxebre: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
@enxebre: Jira Issue OCPBUGS-23466: All pull requests linked via external trackers have merged: Jira Issue OCPBUGS-23466 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/cherrypick release-4.14 |
|
@enxebre: #3218 failed to apply on top of branch "release-4.14": DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
[ART PR BUILD NOTIFIER] This PR has been included in build ose-hypershift-container-v4.15.0-202311221354.p0.g27d0f8f.assembly.stream for distgit hypershift. |
|
Fix included in accepted release 4.14.0-0.nightly-2023-11-23-154014 |
What this PR does / why we need it:
Which issue(s) this PR fixes (optional, use
fixes #<issue_number>(, fixes #<issue_number>, ...)format, where issue_number might be a GitHub issue, or a Jira story:Fixes #
Checklist