Skip to content

agent-installer-api-server to rhel9 - #4659

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.16from
joepvd:agent-installer-rhel9
May 1, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.16from
joepvd:agent-installer-rhel9

Conversation

@joepvd

@joepvd joepvd commented Apr 15, 2024

Copy link
Copy Markdown
Contributor

@joepvd

joepvd commented Apr 15, 2024

Copy link
Copy Markdown
Contributor Author

/hold
Needs to go with the referenced PRs

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 15, 2024
Comment thread images/ose-agent-installer-api-server.yml Outdated

@locriandev locriandev left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@openshift-ci

openshift-ci Bot commented Apr 15, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: locriandev

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 15, 2024
@joepvd
joepvd force-pushed the agent-installer-rhel9 branch from 6314c78 to d12f2c6 Compare April 15, 2024 09:23
@joepvd
joepvd force-pushed the agent-installer-rhel9 branch from d12f2c6 to b71a25a Compare April 23, 2024 10:05
@joepvd joepvd changed the title agent-installer* to rhel9 agent-installer-api-server to rhel9 Apr 23, 2024
@zaneb zaneb mentioned this pull request Apr 24, 2024
@thegreyd

Copy link
Copy Markdown
Contributor

/cherrypick openshift-4.17

@openshift-cherrypick-robot

Copy link
Copy Markdown

@thegreyd: once the present PR merges, I will cherry-pick it on top of openshift-4.17 in a new PR and assign it to you.

Details

In response to this:

/cherrypick openshift-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@thegreyd

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 30, 2024
@joepvd

joepvd commented May 1, 2024

Copy link
Copy Markdown
Contributor Author

/unhold

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label May 1, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit 953d3a7 into openshift-eng:openshift-4.16 May 1, 2024
@openshift-cherrypick-robot

Copy link
Copy Markdown

@thegreyd: new pull request created: #4764

Details

In response to this:

/cherrypick openshift-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants