Skip to content

fix: CRM integration mismatch on callback type - #31880

Merged
kodiakhq[bot] merged 8 commits into
developfrom
fix/livechat-started-crm
Mar 12, 2024
Merged

fix: CRM integration mismatch on callback type#31880
kodiakhq[bot] merged 8 commits into
developfrom
fix/livechat-started-crm

Conversation

@KevLehman

@KevLehman KevLehman commented Mar 4, 2024

Copy link
Copy Markdown
Member

https://rocketchat.atlassian.net/browse/SUP-489

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Mar 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 71664c3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 31 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/apps Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Mar 4, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.47%. Comparing base (d9291b6) to head (71664c3).

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31880      +/-   ##
===========================================
- Coverage    54.55%   54.47%   -0.08%     
===========================================
  Files         2290     2294       +4     
  Lines        50405    50557     +152     
  Branches     10285    10324      +39     
===========================================
+ Hits         27497    27543      +46     
- Misses       20415    20511      +96     
- Partials      2493     2503      +10     
Flag Coverage Δ
e2e 53.51% <ø> (-0.02%) ⬇️
e2e-api 40.05% <ø> (-0.02%) ⬇️
unit 75.45% <100.00%> (-0.59%) ⬇️

Flags with carried forward coverage won't be shown. Click here to find out more.

@KevLehman
KevLehman marked this pull request as ready for review March 5, 2024 14:07
@KevLehman
KevLehman requested a review from a team as a code owner March 5, 2024 14:07
Comment thread .changeset/slow-jobs-think.md Outdated
Comment thread apps/meteor/app/livechat/server/hooks/sendToCRM.ts Outdated
@KevLehman
KevLehman requested a review from a team as a code owner March 11, 2024 14:59
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Mar 12, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 12, 2024
@kodiakhq
kodiakhq Bot merged commit 871efa5 into develop Mar 12, 2024
@kodiakhq
kodiakhq Bot deleted the fix/livechat-started-crm branch March 12, 2024 17:35
@scuciatto scuciatto added this to the 6.7 milestone Mar 12, 2024
gabriellsh added a commit that referenced this pull request Mar 12, 2024
…modalRegions

* 'develop' of github.com:RocketChat/Rocket.Chat: (77 commits)
  fix: CRM integration mismatch on callback type (#31880)
  test: make outgoing integrations fully independent (#31656)
  fix: setDepartment not persisting on chat closed (#31792)
  fix: Trigger app event after inquiry is created (#31909)
  regression: Favorite action announcement (#31940)
  regression: Prevents `ImageGallery` to restore focus on message (#31869)
  i18n: Add missing swedish translation for "Create direct message" key (#31945)
  feat(message-parser): add timestamps pattern (#31810)
  fix: convert network broker stopped lifecycle method to async (#31927)
  regression: undefined guest causing issues (#31932)
  refactor: use uiContexts on livechat (#31919)
  fix: Validate rooms are not taken before processing by queue (#31896)
  fix: livechat sdk reconnect not resubscribing (#31941)
  fix: Quote attachments are not deleted on message update (#31660)
  Revert "regression: Favorite action announcement" (#31935)
  chore: improve room `LeaderBar` behavior with date bubble (#31933)
  regression: Favorite action announcement (#31934)
  chore: Migrate notifyUsersOnMessage to TypeScript (#31877)
  fix: Mentions stop working when mentioned user changes username (#31860)
  regression: `RoomBody` unread and new messages behavior (#31929)
  ...
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants