Skip to content

fix: setDepartment not persisting on chat closed - #31792

Merged
kodiakhq[bot] merged 8 commits into
developfrom
SUP-484
Mar 11, 2024
Merged

fix: setDepartment not persisting on chat closed#31792
kodiakhq[bot] merged 8 commits into
developfrom
SUP-484

Conversation

@MartinSchoeler

@MartinSchoeler MartinSchoeler commented Feb 20, 2024

Copy link
Copy Markdown
Member

SUP-484

Fixes issue of the setDepartment Livechat API method not setting the store value properly (is was only setting on the guest object)

@changeset-bot

changeset-bot Bot commented Feb 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9855bc6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@rocket.chat/livechat Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Feb 20, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.57%. Comparing base (16e6eb5) to head (9855bc6).

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31792      +/-   ##
===========================================
+ Coverage    54.55%   54.57%   +0.02%     
===========================================
  Files         2287     2293       +6     
  Lines        50365    50440      +75     
  Branches     10276    10292      +16     
===========================================
+ Hits         27476    27529      +53     
- Misses       20401    20412      +11     
- Partials      2488     2499      +11     
Flag Coverage Δ
e2e 53.55% <ø> (+0.06%) ⬆️
e2e-api 40.10% <ø> (-0.04%) ⬇️
unit 76.04% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@scuciatto scuciatto added this to the 6.7 milestone Feb 20, 2024
@MartinSchoeler
MartinSchoeler marked this pull request as ready for review February 26, 2024 12:07
@MartinSchoeler
MartinSchoeler requested review from a team as code owners February 26, 2024 12:07
AllanPazRibeiro
AllanPazRibeiro previously approved these changes Mar 4, 2024
MarcosSpessatto
MarcosSpessatto previously approved these changes Mar 4, 2024

@ggazzo ggazzo left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

please provide some test to make sure we never face this again

@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 8, 2024
@MartinSchoeler
MartinSchoeler marked this pull request as draft March 8, 2024 19:40
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 8, 2024
@MartinSchoeler
MartinSchoeler marked this pull request as ready for review March 8, 2024 21:38
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 8, 2024
@MartinSchoeler
MartinSchoeler requested a review from ggazzo March 11, 2024 16:33
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 11, 2024
@kodiakhq
kodiakhq Bot merged commit 9a7d893 into develop Mar 11, 2024
@kodiakhq
kodiakhq Bot deleted the SUP-484 branch March 11, 2024 23:33
gabriellsh added a commit that referenced this pull request Mar 12, 2024
…modalRegions

* 'develop' of github.com:RocketChat/Rocket.Chat: (77 commits)
  fix: CRM integration mismatch on callback type (#31880)
  test: make outgoing integrations fully independent (#31656)
  fix: setDepartment not persisting on chat closed (#31792)
  fix: Trigger app event after inquiry is created (#31909)
  regression: Favorite action announcement (#31940)
  regression: Prevents `ImageGallery` to restore focus on message (#31869)
  i18n: Add missing swedish translation for "Create direct message" key (#31945)
  feat(message-parser): add timestamps pattern (#31810)
  fix: convert network broker stopped lifecycle method to async (#31927)
  regression: undefined guest causing issues (#31932)
  refactor: use uiContexts on livechat (#31919)
  fix: Validate rooms are not taken before processing by queue (#31896)
  fix: livechat sdk reconnect not resubscribing (#31941)
  fix: Quote attachments are not deleted on message update (#31660)
  Revert "regression: Favorite action announcement" (#31935)
  chore: improve room `LeaderBar` behavior with date bubble (#31933)
  regression: Favorite action announcement (#31934)
  chore: Migrate notifyUsersOnMessage to TypeScript (#31877)
  fix: Mentions stop working when mentioned user changes username (#31860)
  regression: `RoomBody` unread and new messages behavior (#31929)
  ...
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants