Skip to content

fix: Language selection box is empty which should be default - #31312

Merged
kodiakhq[bot] merged 3 commits into
RocketChat:developfrom
Sayan4444:nodefaultlang
Dec 26, 2023
Merged

fix: Language selection box is empty which should be default#31312
kodiakhq[bot] merged 3 commits into
RocketChat:developfrom
Sayan4444:nodefaultlang

Conversation

@Sayan4444

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

BEFORE THE FIX
https://github.com/RocketChat/Rocket.Chat/assets/112304873/0fdc1b0f-4e42-4546-87b8-3968e56e1c9c

AFTER THE FIX
https://github.com/RocketChat/Rocket.Chat/assets/112304873/ca4c6cc2-f121-4561-a6ff-5ba1d93a819f

REASON BEHIND CHANGES

The type of 'language' field in default values was string | undefined. When undefined it wont show anything, so we need to use correct type for 'language' which is only 'string' and also if language is undefined(when useUserPreference returns undefiend) we should assign it with blank ''

Issue(s)

This should close issues :-
fix #31307

Steps to test or reproduce

  1. Go to Account
  2. Click on Preferences
  3. See language selection box

Further comments

@Sayan4444
Sayan4444 requested a review from a team as a code owner December 23, 2023 20:03
@changeset-bot

changeset-bot Bot commented Dec 23, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ed0f522

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

dougfabris
dougfabris previously approved these changes Dec 26, 2023
@codecov

codecov Bot commented Dec 26, 2023

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Comparison is base (caa7707) 49.33% compared to head (ed0f522) 49.30%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31312      +/-   ##
===========================================
- Coverage    49.33%   49.30%   -0.03%     
===========================================
  Files         3299     3299              
  Lines        81122    81122              
  Branches     16594    16594              
===========================================
- Hits         40020    40000      -20     
- Misses       36454    36462       +8     
- Partials      4648     4660      +12     
Flag Coverage Δ
e2e 52.52% <ø> (-0.09%) ⬇️
e2e-api 40.63% <ø> (-0.01%) ⬇️
unit 76.39% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dougfabris dougfabris added this to the 6.6 milestone Dec 26, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Dec 26, 2023
@kodiakhq
kodiakhq Bot merged commit 631f6a4 into RocketChat:develop Dec 26, 2023
@Sayan4444
Sayan4444 deleted the nodefaultlang branch December 26, 2023 16:46
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Language selection box is empty which should be default

2 participants