Skip to content

fix: Change Language selection box is empty to Default - #31317

Closed
Lovakush wants to merge 2 commits into
RocketChat:developfrom
Lovakush:lovakush/fix-language-selection-box
Closed

fix: Change Language selection box is empty to Default#31317
Lovakush wants to merge 2 commits into
RocketChat:developfrom
Lovakush:lovakush/fix-language-selection-box

Conversation

@Lovakush

@Lovakush Lovakush commented Dec 25, 2023

Copy link
Copy Markdown

Fixes #31307

@Lovakush
Lovakush requested a review from a team as a code owner December 25, 2023 09:57
@changeset-bot

changeset-bot Bot commented Dec 25, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 49bb984

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@CLAassistant

CLAassistant commented Dec 25, 2023

Copy link
Copy Markdown

CLA assistant check
All committers have signed the CLA.

@Lovakush

Copy link
Copy Markdown
Author

@CLAassistant I have signed CLA now please check my PR

@dougfabris

Copy link
Copy Markdown
Member

Hi there, thanks for the contribution! 🚀 💯

Closing this PR because it's duplicated now and it has been fixed in another one. #31312


Questions? Help needed? Feature Requests?

  • Join our Open Server in the #support channel and feel free to raise a question
  • Join our Community Forum and search/create a post there

@dougfabris dougfabris closed this Dec 26, 2023
@Lovakush
Lovakush deleted the lovakush/fix-language-selection-box branch December 27, 2023 01:24
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Language selection box is empty which should be default

3 participants