Skip to content

Add kubevirt builder image stream - #7579

Closed
dhiller wants to merge 1 commit into
openshift:masterfrom
dhiller:add-kubevirt-builder
Closed

Add kubevirt builder image stream#7579
dhiller wants to merge 1 commit into
openshift:masterfrom
dhiller:add-kubevirt-builder

Conversation

@dhiller

@dhiller dhiller commented Mar 11, 2020

Copy link
Copy Markdown
Contributor

We want to include this image (see #7745) that we can use as builder for our kubevirt stuff here.

@openshift-ci-robot openshift-ci-robot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Mar 11, 2020
@dhiller
dhiller force-pushed the add-kubevirt-builder branch from 79b8ebc to 361992e Compare March 11, 2020 10:45
@openshift-ci-robot openshift-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Mar 11, 2020
@dhiller
dhiller force-pushed the add-kubevirt-builder branch 3 times, most recently from fc0f5c2 to 3045b79 Compare March 18, 2020 08:53
Signed-off-by: Daniel Hiller <daniel.hiller.1972@gmail.com>
@dhiller
dhiller force-pushed the add-kubevirt-builder branch from 3045b79 to a575875 Compare April 22, 2020 12:19
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dhiller

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 22, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dhiller: The following tests failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/rehearse/kubevirt/kubevirt/master/images 3045b79f2ece4d364eafb40cc6414938f824abe3 link /test pj-rehearse
ci/rehearse/kubevirt/kubevirt/master/images-build01 3045b79f2ece4d364eafb40cc6414938f824abe3 link /test pj-rehearse
ci/prow/pj-rehearse a575875 link /test pj-rehearse

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@dhiller

dhiller commented Apr 22, 2020

Copy link
Copy Markdown
Contributor Author

obsolete, see #8211

@dhiller dhiller closed this Apr 22, 2020
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants