Skip to content

Add kubevirt builder image stream - #7745

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
dhiller:add-kubevirt-builder-ci-image
Apr 7, 2020
Merged

Add kubevirt builder image stream#7745
openshift-merge-robot merged 1 commit into
openshift:masterfrom
dhiller:add-kubevirt-builder-ci-image

Conversation

@dhiller

@dhiller dhiller commented Mar 18, 2020

Copy link
Copy Markdown
Contributor

Adds the kubevirt builder image to the supplemental ci images in order to consume it within the build pipeline.

Signed-off-by: Daniel Hiller <daniel.hiller.1972@gmail.com>
@dhiller
dhiller force-pushed the add-kubevirt-builder-ci-image branch from a73c1b6 to 7f2a4c4 Compare April 6, 2020 15:02
@dhiller

dhiller commented Apr 7, 2020

Copy link
Copy Markdown
Contributor Author

/assign @bbguimaraes

Can you PTAL? Thanks!

@bbguimaraes

Copy link
Copy Markdown
Contributor

/lgtm

@bbguimaraes

Copy link
Copy Markdown
Contributor

/retest

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Apr 7, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bbguimaraes, dhiller

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 7, 2020
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-merge-robot
openshift-merge-robot merged commit 94c4069 into openshift:master Apr 7, 2020
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants