Skip to content

Add kubevirt-builder as build base - #8211

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
dhiller:add-kubevirt-builder-to-config
Apr 23, 2020
Merged

Add kubevirt-builder as build base#8211
openshift-merge-robot merged 1 commit into
openshift:masterfrom
dhiller:add-kubevirt-builder-to-config

Conversation

@dhiller

@dhiller dhiller commented Apr 8, 2020

Copy link
Copy Markdown
Contributor

In order to use the kubevirt builder image (provided by this image stream) as base for kubevirt builds we reference it in the configuration.

Goes together with this one: kubevirt/kubevirt#3143

@openshift-ci-robot openshift-ci-robot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Apr 8, 2020
@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 8, 2020
@dhiller
dhiller force-pushed the add-kubevirt-builder-to-config branch from f44807d to ab1a3c3 Compare April 9, 2020 13:23

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

where does this kubevirt-builder container come from? would it not make sense for us to build this build container from the upstream sources?

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

AFAIU there are technical reasons why this had to be done in a separate PR.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

And as you can see here it succeeds in using the image, but fails when trying to use the new script (contained in the related kubevirt PR)

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

ah, i see. makes sense then.

@dhiller
dhiller force-pushed the add-kubevirt-builder-to-config branch from ab1a3c3 to 26428b7 Compare April 22, 2020 12:58
Signed-off-by: Daniel Hiller <daniel.hiller.1972@gmail.com>
@dhiller
dhiller force-pushed the add-kubevirt-builder-to-config branch from 26428b7 to 91ec26d Compare April 22, 2020 13:36
@dhiller
dhiller marked this pull request as ready for review April 22, 2020 14:12
@openshift-ci-robot openshift-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Apr 22, 2020
@dhiller

dhiller commented Apr 23, 2020

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci-robot

openshift-ci-robot commented Apr 23, 2020

Copy link
Copy Markdown
Contributor

@dhiller: The following tests failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/rehearse/kubevirt/kubevirt/master/images ab1a3c30d8e6606d604710ab81929e8b78a1286c link /test pj-rehearse
ci/rehearse/periodic-ci-kubevirt-kubevirt-master-e2e-periodic-gcp-nested-virt 91ec26d link /test pj-rehearse
ci/rehearse/kubevirt/kubevirt/master/e2e-gcp-nested-virt 91ec26d link /test pj-rehearse
ci/prow/pj-rehearse 91ec26d link /test pj-rehearse

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@dhiller

dhiller commented Apr 23, 2020

Copy link
Copy Markdown
Contributor Author

Fix to Dockerfile.ci is here: kubevirt/kubevirt#3326

@davidvossel davidvossel left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Apr 23, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: davidvossel, dhiller

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit adfc6b2 into openshift:master Apr 23, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dhiller: Updated the following 5 configmaps:

  • ci-operator-master-configs configmap in namespace ci at cluster api.ci using the following files:
    • key kubevirt-kubevirt-master.yaml using file ci-operator/config/kubevirt/kubevirt/kubevirt-kubevirt-master.yaml
  • ci-operator-master-configs configmap in namespace ci at cluster app.ci using the following files:
    • key kubevirt-kubevirt-master.yaml using file ci-operator/config/kubevirt/kubevirt/kubevirt-kubevirt-master.yaml
  • ci-operator-master-configs configmap in namespace ci at cluster ci/api-build01-ci-devcluster-openshift-com:6443 using the following files:
    • key kubevirt-kubevirt-master.yaml using file ci-operator/config/kubevirt/kubevirt/kubevirt-kubevirt-master.yaml
  • job-config-master configmap in namespace ci at cluster api.ci using the following files:
    • key kubevirt-kubevirt-master-periodics.yaml using file ci-operator/jobs/kubevirt/kubevirt/kubevirt-kubevirt-master-periodics.yaml
    • key kubevirt-kubevirt-master-presubmits.yaml using file ci-operator/jobs/kubevirt/kubevirt/kubevirt-kubevirt-master-presubmits.yaml
  • job-config-master configmap in namespace ci at cluster app.ci using the following files:
    • key kubevirt-kubevirt-master-periodics.yaml using file ci-operator/jobs/kubevirt/kubevirt/kubevirt-kubevirt-master-periodics.yaml
    • key kubevirt-kubevirt-master-presubmits.yaml using file ci-operator/jobs/kubevirt/kubevirt/kubevirt-kubevirt-master-presubmits.yaml
Details

In response to this:

In order to use the kubevirt builder image (provided by this image stream) as base for kubevirt builds we reference it in the configuration.

Goes together with this one: kubevirt/kubevirt#3143

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants