Skip to content

[openshift-4.17] must-gather inherits from cli-artifacts - #4760

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.17from
openshift-cherrypick-robot:cherry-pick-4754-to-openshift-4.17
Apr 30, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.17from
openshift-cherrypick-robot:cherry-pick-4754-to-openshift-4.17

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #4754

/assign joepvd

@joepvd

joepvd commented Apr 30, 2024

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 30, 2024
@openshift-ci

openshift-ci Bot commented Apr 30, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: joepvd

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 30, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit a6f3a87 into openshift-eng:openshift-4.17 Apr 30, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants