Bump cluster-api to v1.11.1 and support v1beta2 api - #328
Merged
k8s-ci-robot merged 2 commits intoSep 25, 2025
Merged
Conversation
Contributor
Author
|
/ok-to-test |
nunnatsa
force-pushed
the
bump_api-cluster-1.11.1
branch
from
September 18, 2025 11:00
90e431d to
f63e48a
Compare
nunnatsa
force-pushed
the
bump_api-cluster-1.11.1
branch
from
September 25, 2025 04:51
8c739cc to
4dce6f7
Compare
nunnatsa
force-pushed
the
bump_api-cluster-1.11.1
branch
2 times, most recently
from
September 25, 2025 07:22
82b22a4 to
b216c45
Compare
This is a manual cherry pick of sharath-srikanth-chellappa PR 330 Signed-off-by: Nahshon Unna-Tsameret <nunnatsa@redhat.com>
nunnatsa
force-pushed
the
bump_api-cluster-1.11.1
branch
from
September 25, 2025 13:04
57e0036 to
8c61dbf
Compare
Contributor
|
/lgtm |
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: nunnatsa, qinqon The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This was referenced Jul 31, 2026
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
What this PR does / why we need it:
Fixes #321
This patch introduces TLS configuration enforcement for the webhook server in main.go.
Specifically, it ensures that the server uses TLS 1.2 or higher.
Fixes #329
Release notes: