Bump cluster-api to v1.11.1 and support v1beta2 api - #325
changjjjjjjj wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: changjjjjjjj The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
//ok-to-test |
|
/ok-to-test |
|
/override build/coverage |
|
/override "build / coverage" |
|
@changjjjjjjj it seems like the e2e test is broken. I'll try to fix it later today. |
|
/override "build / coverage (pull_request)" |
nunnatsa
left a comment
There was a problem hiding this comment.
Thanks for taking care of this, @changjjjjjjj
Added some inline comments.
| @@ -1,19 +1,17 @@ | |||
| linters-settings: | |||
| ginkgolinter: | |||
There was a problem hiding this comment.
linter-settings option is migrated to settings. I added it again. Thanks
863d649 to
d6a0e15
Compare
|
Hi @nunnatsa, I'm having problem dealing with some e2e test cases.
|
|
Hi @changjjjjjjj - I opened a new PR based on yours, and made some changes to make it work. It's now finally working. Could you please review #328 ? |
Got LGTM. Thanks for your PR and hard work, @changjjjjjjj !! |
|
PR needs rebase. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
@nunnatsa Thanks for the fixes. I will close this PR. |
ACM-26849: fix: mce-29 - Konflux compliance failure
What this PR does / why we need it:
Which issue this PR fixes (optional, in
fixes #<issue number>(, fixes #<issue_number>, ...)format, will close that issue when PR gets merged): fixes #321Special notes for your reviewer:
Release notes: