Skip to content

Enforce TLS v1.2 or higher for webhook server - #330

Closed
sharath-srikanth-chellappa wants to merge 1 commit into
kubernetes-sigs:mainfrom
sharath-srikanth-chellappa:user/sharathsr/enforce-tls-12
Closed

sharath-srikanth-chellappa wants to merge 1 commit into
kubernetes-sigs:mainfrom
sharath-srikanth-chellappa:user/sharathsr/enforce-tls-12

Conversation

@sharath-srikanth-chellappa

Copy link
Copy Markdown
Contributor

What this PR does / why we need it:

Which issue this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close that issue when PR gets merged): fixes #329

This patch introduces TLS configuration enforcement for the webhook server in main.go.
Specifically, it ensures that the server uses TLS 1.2 or higher.

Special notes for your reviewer:

Release notes:

@linux-foundation-easycla

Copy link
Copy Markdown

CLA Missing ID CLA Not Signed

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: sharath-srikanth-chellappa
Once this PR has been reviewed and has the lgtm label, please assign davidvossel for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. label Sep 19, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Welcome @sharath-srikanth-chellappa!

It looks like this is your first PR to kubernetes-sigs/cluster-api-provider-kubevirt 🎉. Please refer to our pull request process documentation to help your PR have a smooth ride to approval.

You will be prompted by a bot to use commands during the review process. Do not be afraid to follow the prompts! It is okay to experiment. Here is the bot commands documentation.

You can also check if kubernetes-sigs/cluster-api-provider-kubevirt has its own contribution guidelines.

You may want to refer to our testing guide if you run into trouble with your tests not passing.

If you are having difficulty getting your pull request seen, please follow the recommended escalation practices. Also, for tips and tricks in the contribution process you may want to read the Kubernetes contributor cheat sheet. We want to make sure your contribution gets all the attention it needs!

Thank you, and welcome to Kubernetes. 😃

@k8s-ci-robot k8s-ci-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 19, 2025
nunnatsa added a commit to nunnatsa/cluster-api-provider-kubevirt that referenced this pull request Sep 25, 2025
This is a manual cherry pick of @sharath-srikanth-chellappa PR kubernetes-sigs#330

Signed-off-by: Nahshon Unna-Tsameret <nunnatsa@redhat.com>
@nunnatsa

Copy link
Copy Markdown
Contributor

Thanks for your PR @sharath-srikanth-chellappa !!

I copied it into #328 and it was just got merged.

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Sep 25, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Enforce TLS v1.2 for the Webhook Server

3 participants