Skip to content

[release/11.0.1xx-preview7] Source code updates from microsoft/vstest - #8065

Merged
nohwnd merged 3 commits into
release/11.0.1xx-preview7from
darc-release/11.0.1xx-preview7-76526043-8cab-46af-952c-5cc1f77261bd
Aug 5, 2026
Merged

[release/11.0.1xx-preview7] Source code updates from microsoft/vstest#8065
nohwnd merged 3 commits into
release/11.0.1xx-preview7from
darc-release/11.0.1xx-preview7-76526043-8cab-46af-952c-5cc1f77261bd

Conversation

@dotnet-maestro

@dotnet-maestro dotnet-maestro Bot commented Jul 30, 2026

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/microsoft/vstest

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/microsoft/vstest:5b0c2fc69de12eb4383a7b6b38acab164a2f68fb..https://github.com/dotnet/dotnet:darc-release/11.0.1xx-preview7-76526043-8cab-46af-952c-5cc1f77261bd

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Important

⚠️ No subscription flowing code in the opposite direction is set up.
Code changes made in the VMR's src/ directory will not flow back to the source repository on branch "rel/18.10".

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 1 pipeline(s).
1 pipeline(s) were filtered out due to trigger conditions.
There may be pipelines that require an authorized user to comment /azp run to run.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team July 30, 2026 17:08

@maestro-approver maestro-approver Bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This pull request contains only source updates from https://github.com/microsoft/vstest.

@dotnet-maestro

dotnet-maestro Bot commented Aug 4, 2026

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@maestro-approver maestro-approver Bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This pull request contains only source updates from https://github.com/microsoft/vstest.

@nohwnd
nohwnd merged commit 79b6dbc into release/11.0.1xx-preview7 Aug 5, 2026
16 checks passed
@nohwnd
nohwnd deleted the darc-release/11.0.1xx-preview7-76526043-8cab-46af-952c-5cc1f77261bd branch August 5, 2026 08:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant