fix: search room not reactive after room name changes - #32123
Merged
Conversation
Contributor
|
Looks like this PR is not ready to merge, because of the following issues:
|
🦋 Changeset detectedLatest commit: 592f496 The changes in this PR will be included in the next version bump. This PR includes changesets to release 31 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAll modified and coverable lines are covered by tests ✅
Additional details and impacted files@@ Coverage Diff @@
## develop #32123 +/- ##
===========================================
- Coverage 54.76% 54.74% -0.03%
===========================================
Files 2299 2302 +3
Lines 50793 50828 +35
Branches 10395 10396 +1
===========================================
+ Hits 27818 27826 +8
- Misses 20471 20495 +24
- Partials 2504 2507 +3
Flags with carried forward coverage won't be shown. Click here to find out more. |
ggazzo
force-pushed
the
test/flaky-edit-name
branch
from
April 3, 2024 21:29
fa284c3 to
a3e8f52
Compare
ggazzo
marked this pull request as ready for review
April 3, 2024 21:31
MartinSchoeler
approved these changes
Apr 3, 2024
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
the key to invalidate clearly only uses the
_id, when the name changes it does not invalidate and keeps rendering the old nameIssue(s)
Steps to test or reproduce
Further comments