Skip to content

feat: Added setting and method to control Livechat system messages visibility - #31751

Merged
kodiakhq[bot] merged 19 commits into
developfrom
feat/livechat-hide-system-messages
Mar 21, 2024
Merged

feat: Added setting and method to control Livechat system messages visibility#31751
kodiakhq[bot] merged 19 commits into
developfrom
feat/livechat-hide-system-messages

Conversation

@aleksandernsilva

@aleksandernsilva aleksandernsilva commented Feb 14, 2024

Copy link
Copy Markdown
Contributor

Do not merge before #30924

Proposed changes (including videos or screenshots)

Issue(s)

PROSVC-49

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Feb 14, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a222f18

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 32 packages
Name Type
@rocket.chat/meteor Minor
@rocket.chat/livechat Minor
@rocket.chat/core-typings Minor
@rocket.chat/rest-typings Minor
@rocket.chat/apps Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Major
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Major
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Major
@rocket.chat/models Patch
@rocket.chat/ui-avatar Major
@rocket.chat/ui-client Major
@rocket.chat/ui-video-conf Major
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Major
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Feb 14, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.72%. Comparing base (d742b25) to head (a222f18).

Additional details and impacted files

Impacted file tree graph

@@           Coverage Diff            @@
##           develop   #31751   +/-   ##
========================================
  Coverage    54.71%   54.72%           
========================================
  Files         2297     2298    +1     
  Lines        50671    50702   +31     
  Branches     10346    10354    +8     
========================================
+ Hits         27725    27747   +22     
- Misses       20455    20460    +5     
- Partials      2491     2495    +4     
Flag Coverage Δ
e2e 53.72% <ø> (-0.05%) ⬇️
e2e-api 40.13% <ø> (+0.11%) ⬆️
unit 75.74% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-hide-system-messages branch from b3a38d9 to 8d1d590 Compare February 14, 2024 18:17
@aleksandernsilva
aleksandernsilva changed the base branch from develop to hooks-ts February 14, 2024 18:22
@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-hide-system-messages branch 2 times, most recently from 59d752a to 7ec7b35 Compare February 16, 2024 18:30
@aleksandernsilva aleksandernsilva changed the title feat: added setting and method to control Livechat system messages visibility feat: Added setting and method to control Livechat system messages visibility Feb 23, 2024
@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-hide-system-messages branch from 97dea59 to 848f778 Compare February 26, 2024 21:41
@aleksandernsilva
aleksandernsilva marked this pull request as ready for review February 27, 2024 18:47
@aleksandernsilva
aleksandernsilva requested review from a team as code owners February 27, 2024 18:47
Comment thread .changeset/cyan-shoes-fly.md Outdated
Comment thread apps/meteor/server/settings/omnichannel.ts Outdated
Comment thread packages/livechat/src/helpers/canRenderMessage.ts
Comment thread packages/livechat/src/widget.ts Outdated
Base automatically changed from hooks-ts to develop February 29, 2024 13:13
@ggazzo
ggazzo dismissed MarcosSpessatto’s stale review February 29, 2024 13:13

The base branch was changed.

@aleksandernsilva
aleksandernsilva marked this pull request as draft February 29, 2024 14:16
@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-hide-system-messages branch from 4d34e6e to 4b120bf Compare February 29, 2024 14:33
@aleksandernsilva
aleksandernsilva marked this pull request as ready for review March 1, 2024 14:19
MarcosSpessatto
MarcosSpessatto previously approved these changes Mar 4, 2024
Comment thread packages/livechat/src/widget.ts
Comment thread apps/meteor/server/settings/omnichannel.ts Outdated
MartinSchoeler
MartinSchoeler previously approved these changes Mar 5, 2024
@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-hide-system-messages branch from 274918d to 00997b0 Compare March 18, 2024 14:33
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: conflict labels Mar 18, 2024
@aleksandernsilva
aleksandernsilva marked this pull request as ready for review March 18, 2024 17:06
@aleksandernsilva aleksandernsilva added this to the 6.7 milestone Mar 18, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 21, 2024
@kodiakhq
kodiakhq Bot merged commit 475c041 into develop Mar 21, 2024
@kodiakhq
kodiakhq Bot deleted the feat/livechat-hide-system-messages branch March 21, 2024 20:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants