Skip to content

feat: added Livechat's new theming settings to Appearance page - #32043

Merged
kodiakhq[bot] merged 7 commits into
developfrom
feat/livechat-theming-appearance
Mar 25, 2024
Merged

feat: added Livechat's new theming settings to Appearance page#32043
kodiakhq[bot] merged 7 commits into
developfrom
feat/livechat-theming-appearance

Conversation

@aleksandernsilva

@aleksandernsilva aleksandernsilva commented Mar 20, 2024

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Added Livechat's new theming settings to Appearance page (available for Premium workspaces)

Newly added settings are:

  • Livechat widget position on the screen: Changes the widget position between left or right of the viewport
  • Livechat background: Changes the message list background. Receives the same value as the CSS's background property.
  • Hide system messages: Changes the visibility of system messages displayed on the widget.
  • Hide "powered by Rocket.Chat": Changes the visibility of Rocket.Chat's watermark on the widget.

Issue(s)

PROSVC-33
PROSVC-78

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Mar 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1af455f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 32 packages
Name Type
@rocket.chat/meteor Minor
@rocket.chat/rest-typings Minor
@rocket.chat/core-services Patch
@rocket.chat/ui-contexts Major
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/ddp-client Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/fuselage-ui-kit Major
@rocket.chat/gazzodown Major
@rocket.chat/livechat Patch
@rocket.chat/ui-avatar Major
@rocket.chat/ui-client Major
@rocket.chat/ui-video-conf Major
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Major
@rocket.chat/core-typings Minor
@rocket.chat/apps Patch
@rocket.chat/cron Patch
@rocket.chat/model-typings Patch
@rocket.chat/license Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/models Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@aleksandernsilva aleksandernsilva added this to the 6.7 milestone Mar 20, 2024
@codecov

codecov Bot commented Mar 20, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.82%. Comparing base (95540f5) to head (1af455f).

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #32043      +/-   ##
===========================================
+ Coverage    52.32%   54.82%   +2.49%     
===========================================
  Files         2177     2299     +122     
  Lines        48729    50739    +2010     
  Branches      9928    10361     +433     
===========================================
+ Hits         25497    27816    +2319     
+ Misses       20954    20436     -518     
- Partials      2278     2487     +209     
Flag Coverage Δ
e2e 53.84% <ø> (+4.84%) ⬆️
e2e-api 40.16% <ø> (+0.03%) ⬆️
unit 75.81% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-theming-appearance branch from 6b5f99a to 9190def Compare March 20, 2024 23:16
@aleksandernsilva aleksandernsilva changed the title feat: added Livechat's new theming settings to appearance page feat: added Livechat's new theming settings to Appearance page Mar 21, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Mar 22, 2024
@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-theming-appearance branch from 9190def to 0bb3afe Compare March 22, 2024 13:52
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 22, 2024
@aleksandernsilva
aleksandernsilva changed the base branch from theming-to-livechat-appearance to develop March 25, 2024 13:19
@aleksandernsilva
aleksandernsilva force-pushed the feat/livechat-theming-appearance branch from 8e65540 to 86e1b01 Compare March 25, 2024 13:20
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 25, 2024
@aleksandernsilva
aleksandernsilva marked this pull request as ready for review March 25, 2024 13:20
@aleksandernsilva
aleksandernsilva requested review from a team as code owners March 25, 2024 13:20
KevLehman
KevLehman previously approved these changes Mar 25, 2024
Comment thread apps/meteor/client/views/omnichannel/appearance/AppearanceFieldLabel.tsx Outdated
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 25, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 25, 2024
@kodiakhq
kodiakhq Bot merged commit d1b1ffe into develop Mar 25, 2024
@kodiakhq
kodiakhq Bot deleted the feat/livechat-theming-appearance branch March 25, 2024 14:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants