Skip to content

fix: use EJSON serializer for data transfer between multi instances - #31540

Merged
kodiakhq[bot] merged 2 commits into
developfrom
fix-multi-instance-data-transfer
Jan 25, 2024
Merged

fix: use EJSON serializer for data transfer between multi instances#31540
kodiakhq[bot] merged 2 commits into
developfrom
fix-multi-instance-data-transfer

Conversation

@sampaiodiego

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@sampaiodiego
sampaiodiego requested a review from a team January 25, 2024 19:41
@changeset-bot

changeset-bot Bot commented Jan 25, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e67011c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Jan 25, 2024

Copy link
Copy Markdown

Codecov Report

Attention: 6 lines in your changes are missing coverage. Please review.

Comparison is base (e72e39e) 49.53% compared to head (e67011c) 49.52%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31540      +/-   ##
===========================================
- Coverage    49.53%   49.52%   -0.02%     
===========================================
  Files         3325     3324       -1     
  Lines        81755    81745      -10     
  Branches     16874    16874              
===========================================
- Hits         40499    40481      -18     
- Misses       36512    36516       +4     
- Partials      4744     4748       +4     
Flag Coverage Δ
e2e 53.21% <ø> (-0.04%) ⬇️
e2e-api 40.61% <0.00%> (-0.01%) ⬇️
unit 76.47% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 25, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 25, 2024
@kodiakhq
kodiakhq Bot merged commit c8ec364 into develop Jan 25, 2024
@kodiakhq
kodiakhq Bot deleted the fix-multi-instance-data-transfer branch January 25, 2024 22:19
@casalsgh casalsgh added this to the 6.6 milestone Jan 26, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants