Skip to content

fix: login buttons remain visible until refresh after disabling authentication service - #31371

Merged
kodiakhq[bot] merged 6 commits into
developfrom
fix/login-buttons-not-hiding
Jan 9, 2024
Merged

fix: login buttons remain visible until refresh after disabling authentication service#31371
kodiakhq[bot] merged 6 commits into
developfrom
fix/login-buttons-not-hiding

Conversation

@pierre-lehnen-rc

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Jan 3, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ea7ab8b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/core-services Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/meteor Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Jan 3, 2024

Copy link
Copy Markdown

Codecov Report

❌ Patch coverage is 25.00000% with 3 lines in your changes missing coverage. Please review.
✅ Project coverage is 49.53%. Comparing base (319f05e) to head (ea7ab8b).
⚠️ Report is 2972 commits behind head on develop.

Additional details and impacted files

Impacted file tree graph

@@           Coverage Diff            @@
##           develop   #31371   +/-   ##
========================================
  Coverage    49.52%   49.53%           
========================================
  Files         3314     3310    -4     
  Lines        81448    81404   -44     
  Branches     16670    16665    -5     
========================================
- Hits         40339    40323   -16     
+ Misses       36397    36368   -29     
- Partials      4712     4713    +1     
Flag Coverage Δ
e2e 53.07% <ø> (+<0.01%) ⬆️
e2e-api 40.66% <25.00%> (+0.01%) ⬆️
unit 76.67% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@pierre-lehnen-rc
pierre-lehnen-rc marked this pull request as ready for review January 5, 2024 20:03
@pierre-lehnen-rc
pierre-lehnen-rc requested review from a team as code owners January 5, 2024 20:03
@pierre-lehnen-rc
pierre-lehnen-rc force-pushed the fix/login-buttons-not-hiding branch from 13c0e86 to 9b0ef66 Compare January 8, 2024 02:12
@pierre-lehnen-rc
pierre-lehnen-rc marked this pull request as draft January 8, 2024 03:06
@pierre-lehnen-rc
pierre-lehnen-rc force-pushed the fix/login-buttons-not-hiding branch from b75dd1a to bbfbeb9 Compare January 8, 2024 03:13
@pierre-lehnen-rc
pierre-lehnen-rc marked this pull request as ready for review January 8, 2024 11:42
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: needs QA labels Jan 9, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 9, 2024
@kodiakhq
kodiakhq Bot merged commit 9a6e9b4 into develop Jan 9, 2024
@kodiakhq
kodiakhq Bot deleted the fix/login-buttons-not-hiding branch January 9, 2024 18:01
gabriellsh added a commit that referenced this pull request Jan 9, 2024
…ve/Notifications

* 'develop' of github.com:RocketChat/Rocket.Chat:
  refactor(client): Move Meteor overrides (#28366)
  fix: login buttons remain visible until refresh after disabling authentication service (#31371)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants