Skip to content

fix: purge messageBox related items from localstorage on logout - #31104

Merged
kodiakhq[bot] merged 3 commits into
developfrom
fix/remove-messageBox-logout
Dec 1, 2023
Merged

fix: purge messageBox related items from localstorage on logout#31104
kodiakhq[bot] merged 3 commits into
developfrom
fix/remove-messageBox-logout

Conversation

@yash-rajpal

@yash-rajpal yash-rajpal commented Nov 29, 2023

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Message Box related items were not being cleared from local storage upon logout

Steps to test or reproduce

Further comments

TC-1043

@yash-rajpal
yash-rajpal requested a review from a team as a code owner November 29, 2023 14:52
@changeset-bot

changeset-bot Bot commented Nov 29, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dbf488c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Nov 29, 2023

Copy link
Copy Markdown

Codecov Report

Merging #31104 (dbf488c) into develop (e57de21) will increase coverage by 1.03%.
The diff coverage is 66.66%.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31104      +/-   ##
===========================================
+ Coverage    56.73%   57.76%   +1.03%     
===========================================
  Files          883     1752     +869     
  Lines        19989    33619   +13630     
  Branches      3688     6874    +3186     
===========================================
+ Hits         11340    19419    +8079     
- Misses        7956    12687    +4731     
- Partials       693     1513     +820     
Flag Coverage Δ
e2e 50.95% <66.66%> (+8.55%) ⬆️
unit 75.53% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Dec 1, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Dec 1, 2023
@kodiakhq
kodiakhq Bot merged commit 5c145e3 into develop Dec 1, 2023
@kodiakhq
kodiakhq Bot deleted the fix/remove-messageBox-logout branch December 1, 2023 17:57
@scuciatto scuciatto added this to the 6.6 milestone Dec 4, 2023
@dionisio-bot dionisio-bot Bot removed the stat: ready to merge PR tested and approved waiting for merge label Dec 4, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants