fix: Visitor message not being sent to webhook - #30933
Merged
kodiakhq[bot] merged 7 commits intoNov 20, 2023
Merged
Conversation
🦋 Changeset detectedLatest commit: 461e17f The changes in this PR will be included in the next version bump. This PR includes changesets to release 30 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
MartinSchoeler
self-requested a review
November 13, 2023 12:50
Codecov Report
Additional details and impacted files@@ Coverage Diff @@
## develop #30933 +/- ##
===========================================
- Coverage 51.42% 51.00% -0.42%
===========================================
Files 814 809 -5
Lines 15046 15175 +129
Branches 2732 2818 +86
===========================================
+ Hits 7737 7740 +3
- Misses 6902 6995 +93
- Partials 407 440 +33
Flags with carried forward coverage won't be shown. Click here to find out more. |
Member
|
Hey! Thanks for your contribution :) Wanna take a look at the contributor license agreement? That's a requirement for all new contributors Follow the instructions from the CLAssistant bot over there, thx! |
…ub.com/ldebowczyk/Rocket.Chat into fix/visitor-message-webhook-omnichanel
Contributor
Author
|
Hi, |
KevLehman
approved these changes
Nov 20, 2023
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
Issue(s)
Steps to test or reproduce
Further comments