Skip to content

fix(meteor): Respect UI_Use_Real_Name setting for reactions - #30069

Merged
ggazzo merged 20 commits into
developfrom
fix/real-name-reactions
Jan 24, 2024
Merged

fix(meteor): Respect UI_Use_Real_Name setting for reactions#30069
ggazzo merged 20 commits into
developfrom
fix/real-name-reactions

Conversation

@yash-rajpal

@yash-rajpal yash-rajpal commented Aug 10, 2023

Copy link
Copy Markdown
Member

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

TC-867

@changeset-bot

changeset-bot Bot commented Aug 10, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7d3a6a4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@yash-rajpal
yash-rajpal marked this pull request as ready for review August 10, 2023 23:36
@yash-rajpal
yash-rajpal requested review from a team as code owners August 10, 2023 23:36
@codecov

codecov Bot commented Aug 10, 2023

Copy link
Copy Markdown

Codecov Report

Attention: 22 lines in your changes are missing coverage. Please review.

Comparison is base (0ed84cb) 59.81% compared to head (7d3a6a4) 49.55%.
Report is 44 commits behind head on develop.

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff              @@
##           develop   #30069       +/-   ##
============================================
- Coverage    59.81%   49.55%   -10.26%     
============================================
  Files         1792     3328     +1536     
  Lines        34893    81801    +46908     
  Branches      7265    16877     +9612     
============================================
+ Hits         20871    40537    +19666     
- Misses       12414    36511    +24097     
- Partials      1608     4753     +3145     
Flag Coverage Δ
e2e 53.24% <4.34%> (+0.12%) ⬆️
e2e-api 40.65% <ø> (?)
unit 76.47% <ø> (-0.05%) ⬇️

Flags with carried forward coverage won't be shown. Click here to find out more.

@scuciatto scuciatto modified the milestone: 6.4.0 Aug 18, 2023
tassoevan
tassoevan previously approved these changes Aug 23, 2023
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Sep 20, 2023
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Oct 11, 2023
@dougfabris dougfabris added this to the 6.5.0 milestone Oct 11, 2023
@dionisio-bot dionisio-bot Bot added stat: needs QA and removed stat: ready to merge PR tested and approved waiting for merge labels Oct 11, 2023
@tassoevan
tassoevan dismissed their stale review October 13, 2023 15:11

We've discussed a more effective strategy to solve this issue without introducing performance drawbacks.

@rodrigok rodrigok modified the milestones: 6.5.0, 6.6.0 Oct 17, 2023
@yash-rajpal
yash-rajpal dismissed stale reviews from tassoevan and gabriellsh via 321eea2 January 22, 2024 20:06
@yash-rajpal
yash-rajpal requested a review from tassoevan January 22, 2024 20:16
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Jan 22, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 24, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 24, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 24, 2024
@scuciatto scuciatto modified the milestones: 6.6, 6.7 Jan 24, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Jan 24, 2024
@ggazzo
ggazzo merged commit 9310e84 into develop Jan 24, 2024
@ggazzo
ggazzo deleted the fix/real-name-reactions branch January 24, 2024 23:20
@scuciatto scuciatto modified the milestones: 6.7, 6.6 Jan 25, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.