Skip to content

Fix run-make/prune-link-args to work with MinGW - #162534

Merged
rust-bors[bot] merged 1 commit into
rust-lang:mainfrom
mati865:prun-link-args-test-mingw
Sep 10, 2026
Merged

Fix run-make/prune-link-args to work with MinGW#162534
rust-bors[bot] merged 1 commit into
rust-lang:mainfrom
mati865:prun-link-args-test-mingw

Conversation

@mati865

@mati865 mati865 commented Sep 9, 2026

Copy link
Copy Markdown
Member

The explanation provided was wrong. Libc doesn't exist on Windows, so naturally this failed on missing lib. I have no idea what shenanigans link.exe does to make this pass, but surely it doesn't actually link libc.

The explanation provided was wrong. Libc doesn't exist on Windows, so
naturally this failed on missing lib. I have no idea what shenanigans
link.exe does to make this pass, but surely it doesn't actually link
libc.
@rustbot rustbot added A-run-make Area: port run-make Makefiles to rmake.rs S-waiting-on-review Status: Awaiting review from the assignee but also interested parties. T-compiler Relevant to the compiler team, which will review and decide on the PR/issue. labels Sep 9, 2026
@rustbot

rustbot commented Sep 9, 2026

Copy link
Copy Markdown
Collaborator

r? @oli-obk

rustbot has assigned @oli-obk.
They will have a look at your PR within the next two weeks and either review your PR or reassign to another reviewer.

Use r? to explicitly pick a reviewer

Why was this reviewer chosen?

The reviewer was selected based on:

  • Owners of files modified in this PR: compiler
  • compiler expanded to 76 candidates
  • Random selection from 21 candidates

@mati865

mati865 commented Sep 9, 2026

Copy link
Copy Markdown
Member Author

@bors try jobs=testmingw,testtest-x86_64-msvc-

rust-bors Bot pushed a commit that referenced this pull request Sep 9, 2026
Fix run-make/prune-link-args to work with MinGW


try-job: test*mingw*
try-job: test*test-x86_64-msvc-*
@rust-bors

This comment has been minimized.

@mati865

mati865 commented Sep 9, 2026

Copy link
Copy Markdown
Member Author

@bors try cancel

@rust-bors

rust-bors Bot commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

Try build cancelled. Cancelled workflows:

Hint: if you want to run another try build, you do not need to manually cancel the previous one. Just run @bors try and bors will cancel the previous build automatically.

@mati865

mati865 commented Sep 9, 2026

Copy link
Copy Markdown
Member Author

@bors try jobs=testmingw,test-x86_64-msvc-*

@rust-bors

This comment has been minimized.

rust-bors Bot pushed a commit that referenced this pull request Sep 9, 2026
Fix run-make/prune-link-args to work with MinGW


try-job: test*mingw*
try-job: test-x86_64-msvc-*
@ChrisDenton

Copy link
Copy Markdown
Member

I have no idea what shenanigans link.exe does to make this pass, but surely it doesn't actually link libc.

I'm pretty sure -lc will just warn about not being a valid option, but not error. So I think it works by accident, essentially.

@rust-bors

rust-bors Bot commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

☀️ Try build successful (CI)
Build commit: b2d472e (b2d472e72faaed3589de78eeb118efcd22197d4c)
Base parent: eca445e (eca445e5ae4a6679cc27d3a09106ce245e13a5a6)

@oli-obk

oli-obk commented Sep 10, 2026

Copy link
Copy Markdown
Contributor

@bors r+ rollup

@rust-bors

rust-bors Bot commented Sep 10, 2026

Copy link
Copy Markdown
Contributor

📌 Commit d9e75ad has been approved by oli-obk

It is now in the queue for this repository.

🌲 The tree is currently closed for pull requests below priority 5. This pull request will be tested once the tree is reopened.

Reason for tree closure: git config problem again

@rust-bors rust-bors Bot added S-waiting-on-bors Status: Waiting on bors to run and complete tests. Bors will change the label on completion. and removed S-waiting-on-review Status: Awaiting review from the assignee but also interested parties. labels Sep 10, 2026
@mati865

mati865 commented Sep 10, 2026

Copy link
Copy Markdown
Member Author

I'm not sure if this is safe to rollup, but should be easy to track down if it causes a failure.

JonathanBrouwer added a commit to JonathanBrouwer/rust that referenced this pull request Sep 10, 2026
… r=oli-obk

Fix run-make/prune-link-args to work with MinGW

The explanation provided was wrong. Libc doesn't exist on Windows, so naturally this failed on missing lib. I have no idea what shenanigans link.exe does to make this pass, but surely it doesn't actually link libc.
rust-bors Bot pushed a commit that referenced this pull request Sep 10, 2026
…uwer

Rollup of 6 pull requests

Successful merges:

 - #162581 (Subtree sync for rustc_codegen_cranelift)
 - #160831 (re-equate the object bound's trait ref with the goal's when considering object bounds)
 - #162057 (Some cleanups around fatal errors and delayed bugs)
 - #162437 (fix(traits): Remove expression references from index suggestions)
 - #162534 (Fix run-make/prune-link-args to work with MinGW)
 - #162576 (doc: Fix external mention of signum() return value regarding NaNs)
rust-bors Bot pushed a commit that referenced this pull request Sep 10, 2026
…uwer

Rollup of 6 pull requests

Successful merges:

 - #162581 (Subtree sync for rustc_codegen_cranelift)
 - #160831 (re-equate the object bound's trait ref with the goal's when considering object bounds)
 - #162057 (Some cleanups around fatal errors and delayed bugs)
 - #162437 (fix(traits): Remove expression references from index suggestions)
 - #162534 (Fix run-make/prune-link-args to work with MinGW)
 - #162576 (doc: Fix external mention of signum() return value regarding NaNs)
rust-bors Bot pushed a commit that referenced this pull request Sep 10, 2026
…uwer

Rollup of 7 pull requests

Successful merges:

 - #162581 (Subtree sync for rustc_codegen_cranelift)
 - #160831 (re-equate the object bound's trait ref with the goal's when considering object bounds)
 - #162057 (Some cleanups around fatal errors and delayed bugs)
 - #162437 (fix(traits): Remove expression references from index suggestions)
 - #162534 (Fix run-make/prune-link-args to work with MinGW)
 - #162554 (Attemped fix for home dir problem on `dist-aarch64-llvm-mingw`)
 - #162576 (doc: Fix external mention of signum() return value regarding NaNs)
@rust-bors
rust-bors Bot merged commit e5d80d1 into rust-lang:main Sep 10, 2026
14 checks passed
@rustbot rustbot added this to the 1.100.0 milestone Sep 10, 2026
rust-bors Bot pushed a commit that referenced this pull request Sep 10, 2026
Rollup merge of #162534 - mati865:prun-link-args-test-mingw, r=oli-obk

Fix run-make/prune-link-args to work with MinGW

The explanation provided was wrong. Libc doesn't exist on Windows, so naturally this failed on missing lib. I have no idea what shenanigans link.exe does to make this pass, but surely it doesn't actually link libc.
@mati865
mati865 deleted the prun-link-args-test-mingw branch September 10, 2026 15:43
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

A-run-make Area: port run-make Makefiles to rmake.rs S-waiting-on-bors Status: Waiting on bors to run and complete tests. Bors will change the label on completion. T-compiler Relevant to the compiler team, which will review and decide on the PR/issue.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants