Skip to content

hypershift: remove non-main previous release e2e jobs - #74416

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
sjenning:rm-prev-release-presub-jobs
Feb 5, 2026
Merged

hypershift: remove non-main previous release e2e jobs#74416
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
sjenning:rm-prev-release-presub-jobs

Conversation

@sjenning

@sjenning sjenning commented Feb 4, 2026

Copy link
Copy Markdown
Contributor

n-1 release e2e jobs only make sense on main for hypershift. main is the only branch where HO changes occur and could cause backward incompatibility with with CPO from the n-1 release.

@openshift-ci
openshift-ci Bot requested review from devguyio and muraee February 4, 2026 21:26
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 4, 2026
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@sjenning: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-hypershift-release-4.21-e2e-aks openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aks-override openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aws openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aws-autonode openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aws-metrics openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aws-minimal openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aws-override openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aws-techpreview openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-aws-upgrade-hypershift-operator openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-azure-aks-ovn-conformance openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-conformance openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-kubevirt-metal-conformance openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-openstack-aws openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-openstack-aws-conformance openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-openstack-aws-csi-cinder openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-openstack-aws-csi-manila openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-e2e-openstack-aws-nfv openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-images openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-okd-scos-e2e-aws-ovn openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-okd-scos-images openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-reqserving-e2e-aws openshift/hypershift presubmit Ci-operator config changed
pull-ci-openshift-hypershift-release-4.21-security openshift/hypershift presubmit Ci-operator config changed

A total of 236 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

@sjenning
sjenning force-pushed the rm-prev-release-presub-jobs branch from f29863a to bb98df3 Compare February 5, 2026 15:37
@enxebre

enxebre commented Feb 5, 2026

Copy link
Copy Markdown
Member

/lgtm

1 similar comment
@bryan-cox

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 5, 2026
@openshift-ci

openshift-ci Bot commented Feb 5, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bryan-cox, enxebre, sjenning

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sjenning

sjenning commented Feb 5, 2026

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Feb 5, 2026
@sjenning

sjenning commented Feb 5, 2026

Copy link
Copy Markdown
Contributor Author

/retest-required

@sjenning sjenning closed this Feb 5, 2026
@sjenning sjenning reopened this Feb 5, 2026
@sjenning

sjenning commented Feb 5, 2026

Copy link
Copy Markdown
Contributor Author

/retest-required

@openshift-ci

openshift-ci Bot commented Feb 5, 2026

Copy link
Copy Markdown
Contributor

@sjenning: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 3200943 into openshift:master Feb 5, 2026
15 checks passed
Sau1506mya pushed a commit to Sau1506mya/release that referenced this pull request Feb 9, 2026
richardsonnick pushed a commit to richardsonnick/release that referenced this pull request Feb 18, 2026
memodi pushed a commit to memodi/release that referenced this pull request Feb 18, 2026
dhensel-rh pushed a commit to dhensel-rh/release that referenced this pull request Feb 19, 2026
rrasouli pushed a commit to rrasouli/release that referenced this pull request Mar 3, 2026
kannon92 pushed a commit to kannon92/release that referenced this pull request Mar 3, 2026
rrasouli pushed a commit to rrasouli/release that referenced this pull request Mar 5, 2026
weinliu pushed a commit to weinliu/release that referenced this pull request Mar 6, 2026
sdodson pushed a commit to sdodson/release that referenced this pull request Mar 8, 2026
tareqalayan pushed a commit to tareqalayan/release that referenced this pull request Mar 13, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants