Skip to content

hypershift: reorder kubevirt presub flow to fix break - #69606

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
sjenning:fix-kubevirt-presub
Sep 23, 2025
Merged

hypershift: reorder kubevirt presub flow to fix break#69606
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
sjenning:fix-kubevirt-presub

Conversation

@sjenning

@sjenning sjenning commented Sep 23, 2025

Copy link
Copy Markdown
Contributor

#68540 regressed the kubevirt presub job because the test cluster is destroyed before the dump stage.

https://prow.ci.openshift.org/job-history/gs/test-platform-results/pr-logs/directory/pull-ci-openshift-hypershift-main-e2e-kubevirt-aws-ovn-reduced

This reorders the flow so that the k8sgpt script no longer errors and we get meaningful dump artifacts from the test cluster.

@openshift-ci
openshift-ci Bot requested review from nirarg and orenc1 September 23, 2025 19:22
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 23, 2025
@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse pull-ci-openshift-hypershift-main-e2e-kubevirt-aws-ovn-reduced

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[REHEARSALNOTIFIER]
@sjenning: the pj-rehearse plugin accommodates running rehearsal tests for the changes in this PR. Expand 'Interacting with pj-rehearse' for usage details. The following rehearsable tests have been affected by this change:

Test name Repo Type Reason
pull-ci-openshift-hypershift-main-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.22-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.21-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.20-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.19-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.18-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.17-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.17-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.16-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.16-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.15-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.15-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.14-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.14-e2e-kubevirt-aws-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-main-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.22-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.21-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.20-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.19-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.18-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.17-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.16-e2e-kubevirt-azure-ovn openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-main-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.22-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Registry content changed
pull-ci-openshift-hypershift-release-4.21-e2e-kubevirt-aws-ovn-reduced openshift/hypershift presubmit Registry content changed

A total of 46 jobs have been affected by this change. The above listing is non-exhaustive and limited to 25 jobs.

A full list of affected jobs can be found here
Prior to this PR being merged, you will need to either run and acknowledge or opt to skip these rehearsals.

Interacting with pj-rehearse

Comment: /pj-rehearse to run up to 5 rehearsals
Comment: /pj-rehearse skip to opt-out of rehearsals
Comment: /pj-rehearse {test-name}, with each test separated by a space, to run one or more specific rehearsals
Comment: /pj-rehearse more to run up to 10 rehearsals
Comment: /pj-rehearse max to run up to 25 rehearsals
Comment: /pj-rehearse auto-ack to run up to 5 rehearsals, and add the rehearsals-ack label on success
Comment: /pj-rehearse list to get an up-to-date list of affected jobs
Comment: /pj-rehearse abort to abort all active rehearsals
Comment: /pj-rehearse network-access-allowed to allow rehearsals of tests that have the restrict_network_access field set to false. This must be executed by an openshift org member who is not the PR author

Once you are satisfied with the results of the rehearsals, comment: /pj-rehearse ack to unblock merge. When the rehearsals-ack label is present on your PR, merge will no longer be blocked by rehearsals.
If you would like the rehearsals-ack label removed, comment: /pj-rehearse reject to re-block merging.

timeout: 1h0m0s
- chain: hypershift-kubevirt-destroy
env:
- default: "/tmp/secret/kubeconfig"

@devguyio devguyio Sep 23, 2025

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

is this env var not needed anymore?

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

That's what the rehearsal is for 😅. Other chains like hypershift-dump do not specify a KUBECONFIG so I think it is already set.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

good point 😁 tagged and 🤞🏽 for the rehearsal

@devguyio

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 23, 2025
@openshift-ci

openshift-ci Bot commented Sep 23, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: devguyio, sjenning

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sjenning

Copy link
Copy Markdown
Contributor Author

/pj-rehearse ack

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@sjenning: now processing your pj-rehearse request. Please allow up to 10 minutes for jobs to trigger or cancel.

@openshift-ci-robot openshift-ci-robot added the rehearsals-ack Signifies that rehearsal jobs have been acknowledged label Sep 23, 2025
@openshift-ci

openshift-ci Bot commented Sep 23, 2025

Copy link
Copy Markdown
Contributor

@sjenning: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 84fce52 into openshift:master Sep 23, 2025
11 checks passed
xiaojiey pushed a commit to xiaojiey/release that referenced this pull request Sep 25, 2025
abhay-nutanix pushed a commit to abhay-nutanix/release that referenced this pull request Oct 16, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. rehearsals-ack Signifies that rehearsal jobs have been acknowledged

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants