Add controller-runtime based tls pkg - #2082
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: damdo The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
6fd675f to
f9272a0
Compare
f9272a0 to
12bddc1
Compare
bertinatto
left a comment
There was a problem hiding this comment.
I'm a bit worried about adding controller-runtime as a dependency, since it would become a transitive dependency for openshift/kubernetes. Is there an alternative place to add this controller?
|
Hey @bertinatto yes this was on my mind when I opened this PR and something @joelanford also brought up. |
|
Adding a dependency on If we need new controllers in |
|
Thanks @p0lyn0mial I've replied to you here: https://redhat-internal.slack.com/archives/C098FU5MRAB/p1768990116222149?thread_ts=1768846936.551799&cid=C098FU5MRAB PTAL |
|
@damdo: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/hold It is very likely we are going to put this in a different repository, we'll discuss and make a decision on this, today at the OCP arch call. |
@damdo any update on ^ ? |
|
Hey @p0lyn0mial yes, we have created the new repo and the PR is now up there: openshift/controller-runtime-common#2 We can close this off now /close |
|
@damdo: Closed this PR. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
DEPRECATED. See openshift/controller-runtime-common#2 instead
--
This PR adds:
pkg/controllerruntimesubfolder for controllerruntime based operators