Skip to content

OCPCLOUD-3349: tls: use centralized TLS profile - #381

Merged
openshift-merge-bot[bot] merged 3 commits into
openshift:mainfrom
damdo:use-centralized-tls-profile
Feb 24, 2026
Merged

OCPCLOUD-3349: tls: use centralized TLS profile#381
openshift-merge-bot[bot] merged 3 commits into
openshift:mainfrom
damdo:use-centralized-tls-profile

Conversation

@damdo

@damdo damdo commented Jan 15, 2026

Copy link
Copy Markdown
Member

Start using centralized TLS profile fetched from the APIServer configuration.

@damdo
damdo requested a review from RadekManak January 15, 2026 20:07
@openshift-ci openshift-ci Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Jan 15, 2026
@openshift-ci
openshift-ci Bot requested review from chrischdi and nrb January 15, 2026 20:10
@damdo

damdo commented Jan 15, 2026

Copy link
Copy Markdown
Member Author

/test unit

@damdo
damdo force-pushed the use-centralized-tls-profile branch from 7cb2875 to 7a3d542 Compare January 16, 2026 10:44
@damdo

damdo commented Jan 16, 2026

Copy link
Copy Markdown
Member Author

/test unit

@damdo damdo changed the title tls: use centralized TLS profile OCPCLOUD-3349: tls: use centralized TLS profile Jan 16, 2026
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jan 16, 2026
@openshift-ci-robot

openshift-ci-robot commented Jan 16, 2026

Copy link
Copy Markdown

@damdo: This pull request references OCPCLOUD-3349 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.22.0" version, but no target version was set.

Details

In response to this:

Start using centralized TLS profile fetched from the APIServer configuration.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@damdo
damdo force-pushed the use-centralized-tls-profile branch from 7a3d542 to a2ff827 Compare January 17, 2026 19:49
@openshift-ci openshift-ci Bot added size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jan 17, 2026
@damdo
damdo force-pushed the use-centralized-tls-profile branch from a2ff827 to 41fb0a3 Compare January 19, 2026 17:46
@openshift-ci openshift-ci Bot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. and removed size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels Jan 19, 2026
@damdo

damdo commented Jan 19, 2026

Copy link
Copy Markdown
Member Author

/assign @JoelSpeed

@damdo

damdo commented Jan 19, 2026

Copy link
Copy Markdown
Member Author

/test unit

@JoelSpeed

Copy link
Copy Markdown
Contributor

Conceptually all looking good here, just need to make sure we understand the direction the centralised TLS configuration group are going

@damdo

damdo commented Jan 20, 2026

Copy link
Copy Markdown
Member Author

Thanks @JoelSpeed I'm planning to move the pkg to a more central location: openshift/library-go#2082

So we can leverage it for other operators.

@openshift-ci openshift-ci Bot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jan 20, 2026
@damdo
damdo force-pushed the use-centralized-tls-profile branch from ee6c698 to ad81545 Compare February 8, 2026 10:27
@damdo

damdo commented Feb 8, 2026

Copy link
Copy Markdown
Member Author

/test unit

@damdo

damdo commented Feb 8, 2026

Copy link
Copy Markdown
Member Author

/test unit e2e-aws-ovn-serial-1of2

@damdo
damdo force-pushed the use-centralized-tls-profile branch from ad81545 to 5e8e1ba Compare February 10, 2026 09:32
@damdo
damdo force-pushed the use-centralized-tls-profile branch from 5e8e1ba to 79cc56c Compare February 14, 2026 09:11
@damdo

damdo commented Feb 17, 2026

Copy link
Copy Markdown
Member Author

/test e2e-aws-ovn-upgrade

@RadekManak RadekManak left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

Comment thread cmd/control-plane-machine-set-operator/main.go
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 17, 2026
@openshift-ci

openshift-ci Bot commented Feb 17, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: RadekManak

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 17, 2026
its use has been removed in favour of the centralized configuration
@openshift-ci openshift-ci Bot added size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. and removed lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Feb 17, 2026
@RadekManak

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 17, 2026
@miyadav

miyadav commented Feb 18, 2026

Copy link
Copy Markdown
Member

/retest-required

1 similar comment
@miyadav

miyadav commented Feb 18, 2026

Copy link
Copy Markdown
Member

/retest-required

@miyadav

miyadav commented Feb 18, 2026

Copy link
Copy Markdown
Member

/verified by @miyadav
test details docs linked in Jira

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Feb 18, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@miyadav: This PR has been marked as verified by @miyadav.

Details

In response to this:

/verified by @miyadav
test details docs linked in Jira

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@damdo

damdo commented Feb 18, 2026

Copy link
Copy Markdown
Member Author

/override ci/prow/e2e-aws-ovn-etcd-scaling

Permafailing

@openshift-ci

openshift-ci Bot commented Feb 18, 2026

Copy link
Copy Markdown
Contributor

@damdo: Overrode contexts on behalf of damdo: ci/prow/e2e-aws-ovn-etcd-scaling

Details

In response to this:

/override ci/prow/e2e-aws-ovn-etcd-scaling

Permafailing

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci

openshift-ci Bot commented Feb 18, 2026

Copy link
Copy Markdown
Contributor

@damdo: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-gcp-ovn-etcd-scaling ee6c698 link false /test e2e-gcp-ovn-etcd-scaling

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@damdo

damdo commented Feb 19, 2026

Copy link
Copy Markdown
Member Author

/test e2e-aws-operator-techpreview

@openshift-merge-bot
openshift-merge-bot Bot merged commit 01ce201 into openshift:main Feb 24, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants