DO NOT MERGE - cpo override PR with 4.16.z testing - #6415
Conversation
Applies the fix delivered with PRs openshift#6393 and openshift#6394 to to OCP versions 4.15.53, 4.15.54 and 4.16.43
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: csrwng The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/test e2e-aws-override |
|
@csrwng: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/close The override test passed for 4.16 🎉 |
|
@csrwng: Closed this PR. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
This is a test-only PR that tests a 4.16.43 ocp release image with an overridden CPO
(Additional testing for #6402)