Skip to content

OCPBUGS-58837: Override CPO image to apply KAS certificate hotfix - #6402

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
csrwng:cert_hotfix_override
Jul 11, 2025
Merged

OCPBUGS-58837: Override CPO image to apply KAS certificate hotfix#6402
openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
csrwng:cert_hotfix_override

Conversation

@csrwng

@csrwng csrwng commented Jul 8, 2025

Copy link
Copy Markdown
Contributor

What this PR does / why we need it:
Applies the fix delivered with PRs #6393 and #6394 to to OCP versions 4.15.53, 4.15.54 and 4.16.43

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #OCPBUGS-58837

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Jul 8, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@csrwng: This pull request references Jira Issue OCPBUGS-58837, which is valid. The bug has been moved to the POST state.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.20.0) matches configured target version for branch (4.20.0)
  • bug is in the state New, which is one of the valid states (NEW, ASSIGNED, POST)

No GitHub users were found matching the public email listed for the QA contact in Jira (yli2@redhat.com), skipping review request.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

What this PR does / why we need it:
Applies the fix delivered with PRs #6393 and #6394 to to OCP versions 4.15.53, 4.15.54 and 4.16.43

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #OCPBUGS-58837

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@csrwng

csrwng commented Jul 8, 2025

Copy link
Copy Markdown
Contributor Author

/area hypershift-operator

@openshift-ci openshift-ci Bot added area/hypershift-operator Indicates the PR includes changes for the hypershift operator and API - outside an OCP release and removed do-not-merge/needs-area labels Jul 8, 2025
@openshift-ci
openshift-ci Bot requested review from rtheis and sjenning July 8, 2025 19:45
@openshift-ci

openshift-ci Bot commented Jul 8, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: csrwng

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 8, 2025
@csrwng

csrwng commented Jul 9, 2025

Copy link
Copy Markdown
Contributor Author

/test e2e-aws-4-19

@csrwng

csrwng commented Jul 9, 2025

Copy link
Copy Markdown
Contributor Author

/override ci/prow/e2e-aws-override

This test is not testing what we need it to test, it needs to be fixed.

@openshift-ci

openshift-ci Bot commented Jul 9, 2025

Copy link
Copy Markdown
Contributor

@csrwng: Overrode contexts on behalf of csrwng: ci/prow/e2e-aws-override

Details

In response to this:

/override ci/prow/e2e-aws-override

This test is not testing what we need it to test, it needs to be fixed.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@csrwng

csrwng commented Jul 9, 2025

Copy link
Copy Markdown
Contributor Author

/retest

@csrwng

csrwng commented Jul 9, 2025

Copy link
Copy Markdown
Contributor Author

/test e2e-aws

@jparrill

Copy link
Copy Markdown
Contributor

/retest

@jparrill

Copy link
Copy Markdown
Contributor

/retest

@sjenning

sjenning commented Jul 10, 2025

Copy link
Copy Markdown
Contributor

Confirmed that override deploys. From CPO pod manifest

https://gcsweb-ci.apps.ci.l2s4.p1.openshiftapps.com/gcs/test-platform-results/pr-logs/pull/openshift_hypershift/6402/pull-ci-openshift-hypershift-main-e2e-aws-override/1943244190586507264/artifacts/e2e-aws-override/hypershift-aws-run-e2e-external/artifacts/TestCreateCluster/namespaces/e2e-clusters-9qdw8-create-cluster-bwkgz/core/pods/control-plane-operator-554f89bd68-xf4hg.yaml

  containers:
    image: quay.io/redhat-user-workloads/crt-redhat-acm-tenant/openshift-cert-hotfix-415-multi:ab45f5df65f19c9e4db4977b0d756c8d87baff0a
...
    env:
...
    - name: CONTROL_PLANE_OPERATOR_IMAGE
      value: quay.io/redhat-user-workloads/crt-redhat-acm-tenant/openshift-cert-hotfix-415-multi:ab45f5df65f19c9e4db4977b0d756c8d87baff0a
    - name: HOSTED_CLUSTER_CONFIG_OPERATOR_IMAGE
      value: quay.io/redhat-user-workloads/crt-redhat-acm-tenant/openshift-cert-hotfix-415-multi:ab45f5df65f19c9e4db4977b0d756c8d87baff0a
    - name: SOCKS5_PROXY_IMAGE
      value: quay.io/redhat-user-workloads/crt-redhat-acm-tenant/openshift-cert-hotfix-415-multi:ab45f5df65f19c9e4db4977b0d756c8d87baff0a
    - name: AVAILABILITY_PROBER_IMAGE
      value: quay.io/redhat-user-workloads/crt-redhat-acm-tenant/openshift-cert-hotfix-415-multi:ab45f5df65f19c9e4db4977b0d756c8d87baff0a
    - name: TOKEN_MINTER_IMAGE
      value: quay.io/redhat-user-workloads/crt-redhat-acm-tenant/openshift-cert-hotfix-415-multi:ab45f5df65f19c9e4db4977b0d756c8d87baff0a
...

Upgrade tested from 4.15.52 to 4.15.53 with CPO override

https://gcsweb-ci.apps.ci.l2s4.p1.openshiftapps.com/gcs/test-platform-results/pr-logs/pull/openshift_hypershift/6402/pull-ci-openshift-hypershift-main-e2e-aws-override/1943244190586507264/artifacts/e2e-aws-override/hypershift-aws-run-e2e-external/artifacts/TestUpgradeControlPlane/namespaces/e2e-clusters-42dkv/hypershift.openshift.io/hostedclusters/control-plane-upgrade-rwg99.yaml

    history:
    - completionTime: "2025-07-10T11:17:42Z"
      image: quay.io/openshift-release-dev/ocp-release:4.15.53-x86_64
      startedTime: "2025-07-10T11:09:42Z"
      state: Completed
      verified: false
      version: 4.15.53
    - completionTime: "2025-07-10T10:52:34Z"
      image: quay.io/openshift-release-dev/ocp-release:4.15.52-x86_64
      startedTime: "2025-07-10T10:31:04Z"
      state: Completed
      verified: false
      version: 4.15.52

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 10, 2025
@csrwng

csrwng commented Jul 11, 2025

Copy link
Copy Markdown
Contributor Author

/retest

csrwng added 2 commits July 11, 2025 15:30
Applies the fix delivered with PRs openshift#6393 and openshift#6394 to
to OCP versions 4.15.53, 4.15.54 and 4.16.43
@csrwng
csrwng force-pushed the cert_hotfix_override branch from 2053e0b to c2a9e7d Compare July 11, 2025 13:30
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Jul 11, 2025
@jparrill

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 11, 2025
@openshift-merge-bot
openshift-merge-bot Bot merged commit 54809d3 into openshift:main Jul 11, 2025
@openshift-ci-robot

Copy link
Copy Markdown

@csrwng: Jira Issue OCPBUGS-58837: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-58837 has been moved to the MODIFIED state.

Details

In response to this:

What this PR does / why we need it:
Applies the fix delivered with PRs #6393 and #6394 to to OCP versions 4.15.53, 4.15.54 and 4.16.43

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #OCPBUGS-58837

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-bot

Copy link
Copy Markdown

[ART PR BUILD NOTIFIER]

Distgit: hypershift
This PR has been included in build ose-hypershift-container-v4.20.0-202507111615.p0.g54809d3.assembly.stream.el9.
All builds following this will include this PR.

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.20.0-0.nightly-2025-09-08-182033

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.20.0-0.nightly-2026-03-14-030631

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.20.0-0.nightly-2026-03-17-142612

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/hypershift-operator Indicates the PR includes changes for the hypershift operator and API - outside an OCP release jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants