Skip to content

OCPBUGS-18326: add Console cap annotation to dashboards - #2254

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
simonpasquier:OCPBUGS-18326
Feb 7, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
simonpasquier:OCPBUGS-18326

Conversation

@simonpasquier

Copy link
Copy Markdown
Contributor

This commit adds the "capability.openshift.io/name: Console" annotation to the dashboard configmaps.

It also renames the existing configmaps from "grafana-xxx" to "xxx" because otherwise CVO would enable the Console capability on all clusters, even those where the capability was initially disabled [1]. The CVO team has a feature request [2] to improve the situation but for now, we have no other option than creating the configmaps with a different name and deleting the old configmaps (anyway the "grafana-" prefix is misleading since OCP doesn't ship any Grafana instance by default).

[1] https://github.com/openshift/enhancements/blob/master/enhancements/installer/component-selection.md#how-to-implement-a-new-capability
[2] https://issues.redhat.com/browse/OTA-1039

  • I added CHANGELOG entry for this change.
  • No user facing changes, so no entry in CHANGELOG was needed.

This commit adds the "capability.openshift.io/name: Console" annotation
to the dashboard configmaps.

It also renames the existing configmaps from "grafana-xxx" to "xxx"
because otherwise CVO would enable the Console capability on all
clusters, even those where the capability was initially disabled [1].
The CVO team has a feature request [2] to improve the situation but for
now, we have no other option than creating the configmaps with a
different name and deleting the old configmaps (anyway the "grafana-"
prefix is misleading since OCP doesn't ship any Grafana instance by
default).

[1] https://github.com/openshift/enhancements/blob/master/enhancements/installer/component-selection.md#how-to-implement-a-new-capability
[2] https://issues.redhat.com/browse/OTA-1039

Signed-off-by: Simon Pasquier <spasquie@redhat.com>
@openshift-ci-robot openshift-ci-robot added jira/severity-low Referenced Jira bug's severity is low for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Feb 5, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@simonpasquier: This pull request references Jira Issue OCPBUGS-18326, which is invalid:

  • expected the bug to target either version "4.16." or "openshift-4.16.", but it targets "4.15.0" instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

This commit adds the "capability.openshift.io/name: Console" annotation to the dashboard configmaps.

It also renames the existing configmaps from "grafana-xxx" to "xxx" because otherwise CVO would enable the Console capability on all clusters, even those where the capability was initially disabled [1]. The CVO team has a feature request [2] to improve the situation but for now, we have no other option than creating the configmaps with a different name and deleting the old configmaps (anyway the "grafana-" prefix is misleading since OCP doesn't ship any Grafana instance by default).

[1] https://github.com/openshift/enhancements/blob/master/enhancements/installer/component-selection.md#how-to-implement-a-new-capability
[2] https://issues.redhat.com/browse/OTA-1039

  • I added CHANGELOG entry for this change.
  • No user facing changes, so no entry in CHANGELOG was needed.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@simonpasquier

Copy link
Copy Markdown
Contributor Author

/jira refresh

@openshift-ci-robot openshift-ci-robot added jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. and removed jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Feb 5, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@simonpasquier: This pull request references Jira Issue OCPBUGS-18326, which is valid. The bug has been moved to the POST state.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.16.0) matches configured target version for branch (4.16.0)
  • bug is in the state ASSIGNED, which is one of the valid states (NEW, ASSIGNED, POST)

Requesting review from QA contact:
/cc @juzhao

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from jan--f, juzhao and slashpai February 5, 2024 15:30
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 5, 2024
@openshift-ci

openshift-ci Bot commented Feb 5, 2024

Copy link
Copy Markdown
Contributor

@simonpasquier: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/versions d84f0f3 link false /test versions
ci/prow/e2e-aws-ovn-single-node d84f0f3 link false /test e2e-aws-ovn-single-node

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@jan--f

jan--f commented Feb 7, 2024

Copy link
Copy Markdown
Contributor

Nice, thank you! Love the commit message ❤️

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 7, 2024
@simonpasquier

Copy link
Copy Markdown
Contributor Author

/skip

@machine424

Copy link
Copy Markdown
Contributor

Thanks
/lgtm

@openshift-ci

openshift-ci Bot commented Feb 7, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jan--f, machine424, simonpasquier

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [jan--f,machine424,simonpasquier]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit dbcfed9 into openshift:master Feb 7, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@simonpasquier: Jira Issue OCPBUGS-18326: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-18326 has been moved to the MODIFIED state.

Details

In response to this:

This commit adds the "capability.openshift.io/name: Console" annotation to the dashboard configmaps.

It also renames the existing configmaps from "grafana-xxx" to "xxx" because otherwise CVO would enable the Console capability on all clusters, even those where the capability was initially disabled [1]. The CVO team has a feature request [2] to improve the situation but for now, we have no other option than creating the configmaps with a different name and deleting the old configmaps (anyway the "grafana-" prefix is misleading since OCP doesn't ship any Grafana instance by default).

[1] https://github.com/openshift/enhancements/blob/master/enhancements/installer/component-selection.md#how-to-implement-a-new-capability
[2] https://issues.redhat.com/browse/OTA-1039

  • I added CHANGELOG entry for this change.
  • No user facing changes, so no entry in CHANGELOG was needed.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build cluster-monitoring-operator-container-v4.16.0-202402071412.p0.gdbcfed9.assembly.stream.el9 for distgit cluster-monitoring-operator.
All builds following this will include this PR.

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.16.0-0.nightly-2024-02-08-025333

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/severity-low Referenced Jira bug's severity is low for the branch this PR is targeting. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants