Skip to content

OCPBUGS-18326: revert previous fix to avoid activating console capability during update - #2118

Merged
openshift-ci[bot] merged 1 commit into
openshift:masterfrom
raptorsun:OCPBUGS-18326-revert
Oct 16, 2023
Merged

openshift-ci[bot] merged 1 commit into
openshift:masterfrom
raptorsun:OCPBUGS-18326-revert

Conversation

@raptorsun

Copy link
Copy Markdown
Contributor

Revert PR#2101
As Trevor commented in https://issues.redhat.com/browse/OCPBUGS-18326 , the capability annotation will activate the required capability implicitly on clusters during an update. We should avoid activating capabilities during update.

@openshift-ci-robot openshift-ci-robot added jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Oct 13, 2023
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@raptorsun: This pull request references Jira Issue OCPBUGS-18326, which is valid. The bug has been moved to the POST state.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.15.0) matches configured target version for branch (4.15.0)
  • bug is in the state ASSIGNED, which is one of the valid states (NEW, ASSIGNED, POST)

Requesting review from QA contact:
/cc @juzhao

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Revert PR#2101
As Trevor commented in https://issues.redhat.com/browse/OCPBUGS-18326 , the capability annotation will activate the required capability implicitly on clusters during an update. We should avoid activating capabilities during update.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 13, 2023
@raptorsun

Copy link
Copy Markdown
Contributor Author

/close
Will look for better solutions in the 4.15 development cycle.

@openshift-ci openshift-ci Bot closed this Oct 13, 2023
@openshift-ci

openshift-ci Bot commented Oct 13, 2023

Copy link
Copy Markdown
Contributor

@raptorsun: Closed this PR.

Details

In response to this:

/close
Will look for better solutions in the 4.15 development cycle.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@raptorsun: This pull request references Jira Issue OCPBUGS-18326. The bug has been updated to no longer refer to the pull request using the external bug tracker.

Details

In response to this:

Revert PR#2101
As Trevor commented in https://issues.redhat.com/browse/OCPBUGS-18326 , the capability annotation will activate the required capability implicitly on clusters during an update. We should avoid activating capabilities during update.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@raptorsun
raptorsun deleted the OCPBUGS-18326-revert branch October 13, 2023 09:02
@raptorsun
raptorsun restored the OCPBUGS-18326-revert branch October 13, 2023 09:04
@raptorsun

Copy link
Copy Markdown
Contributor Author

/reopen

@openshift-ci openshift-ci Bot reopened this Oct 13, 2023
@openshift-ci

openshift-ci Bot commented Oct 13, 2023

Copy link
Copy Markdown
Contributor

@raptorsun: Reopened this PR.

Details

In response to this:

/reopen

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@raptorsun: This pull request references Jira Issue OCPBUGS-18326, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.15.0) matches configured target version for branch (4.15.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

Requesting review from QA contact:
/cc @juzhao

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Revert PR#2101
As Trevor commented in https://issues.redhat.com/browse/OCPBUGS-18326 , the capability annotation will activate the required capability implicitly on clusters during an update. We should avoid activating capabilities during update.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@simonpasquier

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Oct 13, 2023
@openshift-ci

openshift-ci Bot commented Oct 13, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: raptorsun, simonpasquier

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [raptorsun,simonpasquier]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@raptorsun

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci

openshift-ci Bot commented Oct 16, 2023

Copy link
Copy Markdown
Contributor

@raptorsun: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-ci
openshift-ci Bot merged commit 1929183 into openshift:master Oct 16, 2023
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@raptorsun: Jira Issue OCPBUGS-18326: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-18326 has been moved to the MODIFIED state.

Details

In response to this:

Revert PR#2101
As Trevor commented in https://issues.redhat.com/browse/OCPBUGS-18326 , the capability annotation will activate the required capability implicitly on clusters during an update. We should avoid activating capabilities during update.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.16.0-0.nightly-2024-02-08-025333

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants