e2e-tests: ensure pod assertions find at least one pod - #2028
Conversation
|
/retest-required |
|
/retest |
| } | ||
|
|
||
| if len(pods.Items) == 0 { | ||
| return fmt.Errorf("%w - failed to find pods in %s matching %s", err, namespace, labelSelector) |
There was a problem hiding this comment.
err should be nil
| return fmt.Errorf("%w - failed to find pods in %s matching %s", err, namespace, labelSelector) | |
| return fmt.Errorf("failed to find pods in %s matching %s", namespace, labelSelector) |
abb5749 to
250cfae
Compare
250cfae to
18b5914
Compare
|
/unhold |
Signed-off-by: Sunil Thaha <sthaha@redhat.com>
18b5914 to
cca89e0
Compare
|
@sthaha: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: marioferh, simonpasquier, sthaha The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
No description provided.