OCPBUGS-14922: skip console-plugin installation if console CO is absent - #2011
openshift-merge-robot merged 1 commit into
Conversation
|
@sthaha: This pull request references Jira Issue OCPBUGS-14922, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
66211b3 to
58de492
Compare
|
/jira refresh |
|
@jan--f: This pull request references Jira Issue OCPBUGS-14922, which is valid. The bug has been moved to the POST state. 3 validation(s) were run on this bug
Requesting review from QA contact: DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
681364c to
934c4df
Compare
8f03f88 to
2910eb5
Compare
simonpasquier
left a comment
There was a problem hiding this comment.
The handleEvent() function needs to react on changes to the clusterversion object.
cluster-monitoring-operator/pkg/operator/operator.go
Lines 568 to 596 in 8d67591
I'd also replace the existing if blocks by type switches (e.g. https://go.dev/tour/methods/16).
e67e7a1 to
f7f1f87
Compare
|
/retest-required |
… is disabled This PR fixes the failure to upgrade from OpenShift version 4.13 to 4.14 in environments (such as telco) where `console` capability is disabled. The upgrade fails when CMO executes the console-plugin task which fails due to missing ConsolePlugin CRD. The fix checks if `console` capability is enabled in ClusterVersion - version and bails out (with a log message) if it is disabled. Signed-off-by: Sunil Thaha <sthaha@redhat.com>
f7f1f87 to
3aac627
Compare
|
/retest-required |
1 similar comment
|
/retest-required |
|
@sthaha: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
/retest-required |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jan--f, sthaha The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
@sthaha: Jira Issue OCPBUGS-14922: All pull requests linked via external trackers have merged: Jira Issue OCPBUGS-14922 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
This PR fixes the failure to upgrade from OpenShift version 4.13 to 4.14 in environments (such as telco) where
consoleClusterOperator is disabled. The upgrade fails happens when CMO executes the console-plugin task which fails due to missing ConsolePlugin.The fix checks for the presence of
consoleClusterOperator and bails out (with a log message) if it is absent.