Skip to content

AGENT-1511: Support multi node HA ( 3/4/5 control planes) and SNO Cluster Topology for OVE NoRegistry - #1884

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
pawanpinjarkar:5-control-plane-node-support-4-ove
May 7, 2026
Merged

AGENT-1511: Support multi node HA ( 3/4/5 control planes) and SNO Cluster Topology for OVE NoRegistry#1884
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
pawanpinjarkar:5-control-plane-node-support-4-ove

Conversation

@pawanpinjarkar

@pawanpinjarkar pawanpinjarkar commented Apr 22, 2026

Copy link
Copy Markdown
Contributor

Support multi node HA ( 3/4/5 control planes) ans SNO Cluster Topology for OVE NoRegistry Flow in dev-scripts and UI Automation

  • Update logic to support multi node HA ( 3/4/5 control planes) and SNO topology for the ISO_NO_REGISTRY NoRegistryClusterInstall
  • Update the automated UI testing script to handle the cluster configuration.
  • Increase MASTER_DISK for 4 and 5 control planes

@openshift-ci
openshift-ci Bot requested review from celebdor and lranjbar April 22, 2026 19:21
Comment thread agent/01_agent_requirements.sh Outdated
@pawanpinjarkar

Copy link
Copy Markdown
Contributor Author

@openshift-ci
openshift-ci Bot requested review from andfasano, bfournie, rwsu and sadasu and removed request for celebdor and lranjbar April 22, 2026 19:22
}

menuItem.MustClick()
time.Sleep(500 * time.Millisecond)

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Q: why all these sleeps? IIRC there was a default way in go-rod to "slow down" a little bit the operations

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

That was the existing pattern used but I agree to remove time.sleep and use the go-rod's supported functions instead. I will open a new follow up PR for fixing the existing ones.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

to be clear, I remove the time.sleep added originally in this PR. There are still existing time.sleep() calls in the automation script which I will remove gradually in a follow up PR as an improvement.
cc @andfasano @bfournie

Comment thread agent/isobuilder/ui_driven_cluster_installation/main.go Outdated
@pawanpinjarkar pawanpinjarkar changed the title AGENT-1511: Support HA ( 5 control planes) Cluster Topology for OVE NoRegistry AGENT-1511: Support multi node HA ( 3/4/5 control planes) Cluster Topology for OVE NoRegistry Apr 27, 2026
Comment thread agent/isobuilder/ui_driven_cluster_installation/main.go Outdated
Comment thread agent/isobuilder/ui_driven_cluster_installation/main.go Outdated
Comment thread agent/isobuilder/ui_driven_cluster_installation/main.go Outdated
Comment thread agent/isobuilder/ui_driven_cluster_installation/main.go
@pawanpinjarkar
pawanpinjarkar force-pushed the 5-control-plane-node-support-4-ove branch from 9e5352b to 5e3bcc7 Compare May 5, 2026 13:32
@pawanpinjarkar
pawanpinjarkar requested a review from rwsu May 5, 2026 15:27
@pawanpinjarkar
pawanpinjarkar force-pushed the 5-control-plane-node-support-4-ove branch from 5e3bcc7 to f6e9598 Compare May 5, 2026 21:21
@pawanpinjarkar pawanpinjarkar changed the title AGENT-1511: Support multi node HA ( 3/4/5 control planes) Cluster Topology for OVE NoRegistry AGENT-1511: Support multi node HA ( 3/4/5 control planes) and SNO Cluster Topology for OVE NoRegistry May 7, 2026
@pawanpinjarkar
pawanpinjarkar force-pushed the 5-control-plane-node-support-4-ove branch from f6e9598 to 5e42e20 Compare May 7, 2026 11:18
@pawanpinjarkar

Copy link
Copy Markdown
Contributor Author

I have verified it locally and the UI automation correctly sets the 5 control planes when AGENT_E2E_TEST_SCENARIO=5CONTROL_IPV4

@openshift-ci

openshift-ci Bot commented May 7, 2026

Copy link
Copy Markdown

@pawanpinjarkar: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-metal-ipi-virtualmedia 5e42e20 link false /test e2e-metal-ipi-virtualmedia

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@andfasano

Copy link
Copy Markdown
Member

/approve

@pawanpinjarkar can you please run an iso-no-registry job?

@openshift-ci

openshift-ci Bot commented May 7, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: andfasano

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 7, 2026
@pawanpinjarkar

Copy link
Copy Markdown
Contributor Author

/approve

@pawanpinjarkar can you please run an iso-no-registry job?

Hi @andfasano the compact job was run and passed ci/prow/e2e-agent-compact-ipv4-iso-no-registry

@sadasu

sadasu commented May 7, 2026

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 7, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 59e7572 into openshift-metal3:master May 7, 2026
17 of 18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants