Skip to content

[ART-10306] [ose-frr] set for_payload as true - #5209

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.17from
ashwindasr:ose-frr-release-true
Aug 8, 2024
Merged

[ART-10306] [ose-frr] set for_payload as true#5209
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:openshift-4.17from
ashwindasr:ose-frr-release-true

Conversation

@ashwindasr

Copy link
Copy Markdown
Contributor

@ashwindasr
ashwindasr force-pushed the ose-frr-release-true branch from 10e17c1 to 35c1904 Compare August 7, 2024 19:29
Comment thread images/ose-frr.yml
for_payload: false
for_payload: true
dependents:
- ose-metallb-operator

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

we would also need to add cno as dependent i think

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

cluster-network-operator right?

@thegreyd thegreyd Aug 7, 2024

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

hmm wait i'm wrong, dependent is a child image. but cno is not child image of frr

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

parent/child isn't the best way to think about this.. I'll wait for confirmation

@Ximinhan Ximinhan Aug 8, 2024

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

currently the dependents is more used for olm, the cluster-network-operator don't have olm yet, so it won't effect for now, we no need to add cluster-network-operator to dependents for now

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks ximin, updated

@ashwindasr
ashwindasr force-pushed the ose-frr-release-true branch 2 times, most recently from 4f88da5 to dbdf9fb Compare August 7, 2024 19:45
@ashwindasr

Copy link
Copy Markdown
Contributor Author

@ashwindasr

Copy link
Copy Markdown
Contributor Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 7, 2024
@ashwindasr
ashwindasr force-pushed the ose-frr-release-true branch from dbdf9fb to 7598857 Compare August 8, 2024 13:34
@ashwindasr

Copy link
Copy Markdown
Contributor Author

/unhold

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 8, 2024
@thegreyd

thegreyd commented Aug 8, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 8, 2024
@ashwindasr

Copy link
Copy Markdown
Contributor Author

/approve

@openshift-ci

openshift-ci Bot commented Aug 8, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ashwindasr

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 8, 2024
@ashwindasr

Copy link
Copy Markdown
Contributor Author

/cherry-pick openshift-4.18

@openshift-cherrypick-robot

Copy link
Copy Markdown

@ashwindasr: once the present PR merges, I will cherry-pick it on top of openshift-4.18 in a new PR and assign it to you.

Details

In response to this:

/cherry-pick openshift-4.18

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 31d41c5 into openshift-eng:openshift-4.17 Aug 8, 2024
@openshift-cherrypick-robot

Copy link
Copy Markdown

@ashwindasr: new pull request created: #5218

Details

In response to this:

/cherry-pick openshift-4.18

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants