Skip to content

fix: tiny tree downstream import issues - #7355

Merged
sidharthv96 merged 2 commits into
masterfrom
sidv/fixTinyTreeLayout
Jan 29, 2026
Merged

sidharthv96 merged 2 commits into
masterfrom
sidv/fixTinyTreeLayout

Conversation

@sidharthv96

Copy link
Copy Markdown
Member

📑 Summary

Move non-layered-tidy-tree-layout to devDependencies
This allows esbuild to inline it during build, which fixes import issues caused by ESM vs CJS.

📏 Design Decisions

Describe the way your implementation works or what design decisions you made if applicable.

📋 Tasks

Make sure you

  • 📖 have read the contribution guidelines
  • 💻 have added necessary unit/e2e tests.
  • 📓 have added documentation. Make sure MERMAID_RELEASE_VERSION is used for all new features.
  • 🦋 If your PR makes a change that should be noted in one or more packages' changelogs, generate a changeset by running pnpm changeset and following the prompts. Changesets that add features should be minor and those that fix bugs should be patch. Please prefix changeset messages with feat:, fix:, or chore:.

sidharthv96 and others added 2 commits January 29, 2026 14:01
This allows esbuild to inline it during build, which fixes import issues caused by ESM vs CJS.

Co-authored-by: Alois Klink <alois@aloisklink.com>
@changeset-bot

changeset-bot Bot commented Jan 29, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0f94b84

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@mermaid-js/layout-tidy-tree Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@sidharthv96
sidharthv96 merged commit 57645cf into master Jan 29, 2026
16 of 19 checks passed
@sidharthv96
sidharthv96 deleted the sidv/fixTinyTreeLayout branch January 29, 2026 08:43
@argos-ci

argos-ci Bot commented Jan 29, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Argos notifications ↗︎

Build Status Details Updated (UTC)
default (Inspect) ✅ No changes detected - Jan 29, 2026, 8:46 AM

@github-actions github-actions Bot mentioned this pull request Jan 29, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant