Repository navigation
Fix edge labels for ER and requirement diagrams when config.flowchart.htmlLabels is false - #7296
Conversation
…art: { htmlLabels: false }
on-behalf-of: @Mermaid-Chart <hello@mermaidchart.com>
🦋 Changeset detectedLatest commit: aac86f7 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
✅ Deploy Preview for mermaid-js ready!
To edit notification comments on pull requests, go to your Netlify project configuration. |
@mermaid-js/examples
mermaid
@mermaid-js/layout-elk
@mermaid-js/layout-tidy-tree
@mermaid-js/mermaid-zenuml
@mermaid-js/parser
@mermaid-js/tiny
commit: |
Codecov Report❌ Patch coverage is Additional details and impacted files@@ Coverage Diff @@
## develop #7296 +/- ##
==========================================
- Coverage 3.57% 3.57% -0.01%
==========================================
Files 473 474 +1
Lines 47475 47498 +23
Branches 734 734
==========================================
Hits 1696 1696
- Misses 45779 45802 +23
Flags with carried forward coverage won't be shown. Click here to find out more.
🚀 New features to boost your workflow:
|
|
The latest updates on your projects. Learn more about Argos notifications ↗︎
|
aloisklink
left a comment
There was a problem hiding this comment.
LGTM!
@darshanr0107, can you maybe add a before and after screenshot to the PR description? After that, I think it's good to merge.
config.flowchart.htmlLabels is false
📑 Summary
This PR fixes edge label rendering issues in ER and requirement diagrams when
flowchart.htmlLabelsis set tofalse.Note: This PR should be merged before #6995
Before

After

📏 Design Decisions
Describe the way your implementation works or what design decisions you made if applicable.
📋 Tasks
Make sure you
MERMAID_RELEASE_VERSIONis used for all new features.pnpm changesetand following the prompts. Changesets that add features should beminorand those that fix bugs should bepatch. Please prefix changeset messages withfeat:,fix:, orchore:.