Skip to content

[WIP] Added OWNERS for localization projects - #14108

Closed
rlenferink wants to merge 1 commit into
kubernetes:masterfrom
rlenferink:add-i18n-owners
Closed

rlenferink wants to merge 1 commit into
kubernetes:masterfrom
rlenferink:add-i18n-owners

Conversation

@rlenferink

Copy link
Copy Markdown
Member

Added sig-docs-l10n-owners to the OWNERS_ALIASES and added a new OWNERS file to the i18n directory. This will enable localization teams to approve changes for content in the /i18n/ directory.

xref. #14048
/cc @zacharysarah
/cc @kubernetes/sig-docs-l10n-admins

@k8s-ci-robot
k8s-ci-robot requested review from a team and zacharysarah April 30, 2019 17:04
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Apr 30, 2019
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
To fully approve this pull request, please assign additional approvers.
We suggest the following additional approver: ryanmcginnis

If they are not already assigned, you can assign the PR to them by writing /assign @ryanmcginnis in a comment when ready.

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the sig/docs Categorizes an issue or PR as relevant to SIG Docs. label Apr 30, 2019
Comment thread OWNERS_ALIASES Outdated
@rlenferink

rlenferink commented Apr 30, 2019

Copy link
Copy Markdown
Member Author

Hugo still returns the following:

ERROR 2019/04/30 18:02:43 "/src/i18n/OWNERS:1:1": failed to load translations: failed to unmarshal OWNERS: unsupported file extension

It still looks like the problem for ignoring additional files hasn't been solved (#10888). I'll have a look into this shortly.

@rlenferink rlenferink changed the title Added OWNERS for localization projects [WIP] Added OWNERS for localization projects Apr 30, 2019
@k8s-ci-robot k8s-ci-robot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Apr 30, 2019
@zacharysarah

Copy link
Copy Markdown
Contributor

It still looks like the problem for ignoring additional files hasn't be solved (#10888).

Aw, shoot--I forgot about that.

@irvifa

irvifa commented May 2, 2019

Copy link
Copy Markdown
Member

So this still can't be used when specifying OWNER_ALIASES?

@rlenferink

Copy link
Copy Markdown
Member Author

So this still can't be used when specifying OWNER_ALIASES?

Unfortunately, no, not at the moment. I created the following issue gohugoio/hugo#5922 which maybe will clarify some things for us.

@rlenferink

Copy link
Copy Markdown
Member Author

There are 2 options to have this supported:

Option 1:
Update Hugo to respect the ignoreFiles specified in the config.toml

Option 2:
Update our approve plugin to support approve filters. Filters are already supported for labels (documented here), but support for approvers has not been implemented yet.

We could have something as the following in our OWNERS file:

reviewers:
- sig-docs-en-reviews

approvers:
- sig-docs-en-owners

labels:
- sig/docs

filters:
  "^i18n\/.*":
    approvers:
    - sig-docs-l10n-owners
  "^README-.*\.md$":
    approvers:
    - sig-docs-l10n-owners

This will make sure all files in the i18n/ directory match and it will match the README-*.md files as well.

kubernetes/test-infra#7690 is created for additional information and tracking progress.

@nikhita

nikhita commented May 4, 2019

Copy link
Copy Markdown
Member

👋 just wanted to chime in about the owners file...

re option 2: if you are using filters, you can't use reviewers, approvers and labels field outside of filters. Also, regular expressions won't work with the approve plugin (like @rlenferink mentioned).

You can do something like https://github.com/kubernetes/kubernetes/blob/master/OWNERS though i.e. if you want, you can add regex for labels or required_reviewers. The only regex that will work with approvers is * today.

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label May 22, 2019
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@rlenferink: PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@zacharysarah

Copy link
Copy Markdown
Contributor

@rlenferink 👋 Given that Hugo still doesn't follow its own ignore rules, I'm going to close this PR. Please feel free to reopen it if the situation changes.

/close

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@zacharysarah: Closed this PR.

Details

In response to this:

@rlenferink 👋 Given that Hugo still doesn't follow its own ignore rules, I'm going to close this PR. Please feel free to reopen it if the situation changes.

/close

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. sig/docs Categorizes an issue or PR as relevant to SIG Docs. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants