[WIP] Add i18n/OWNERS, update aliases - #10888
zacharysarah wants to merge 3 commits into
Conversation
|
Deploy preview for kubernetes-io-master-staging failed. Built with commit d27ae45 https://app.netlify.com/sites/kubernetes-io-master-staging/deploys/5be0a4090ff8d40b16b2086b |
|
Deploy preview for kubernetes-io-master-staging failed. Built with commit d9d89a3 https://app.netlify.com/sites/kubernetes-io-master-staging/deploys/5be0a606daeb0a33461d9a77 |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: If they are not already assigned, you can assign the PR to them by writing The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
After testing, it looks like the build failure is due specifically to the presence of an OWNERS file in 12:20:43 PM: Executing user command: hugo --enableGitInfo -b $DEPLOY_PRIME_URL
12:20:43 PM: Error: Failed to load translations in file "OWNERS": failed to unmarshal OWNERS: unsupported file extensionThe top-level ignoreFiles = [ "^OWNERS$", "README[-]+[a-z]*\\.md", "^node_modules$", "content/en/docs/doc-contributor-tools" ]So I'm guessing that there's some kind of content enforcement for https://discourse.gohugo.io/t/ignorefiles-and-i18n-expected-behavior/15097 |
Re-attempt #10811
/assign @cody-clark @chenopis