Skip to content

test: update test certs and k8s tls secret - #203

Merged
k8s-ci-robot merged 1 commit into
kubernetes-csi:mainfrom
iPraveenParihar:test/update-certs-and-tls-secret
Jan 30, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes-csi:mainfrom
iPraveenParihar:test/update-certs-and-tls-secret

Conversation

@iPraveenParihar

Copy link
Copy Markdown
Contributor

This commit updates certs and k8s tls secret used for e2e test with expiry set to 365 days.
Earlier certs had validity of 60 days which were expired.

What type of PR is this?

Uncomment only one /kind <> line, hit enter to put that in a new line, and remove leading whitespaces from that line:

/kind api-change
/kind bug
/kind cleanup
/kind design
/kind documentation

/kind failing-test

/kind feature
/kind flake

What this PR does / why we need it:

This PR updates certs and k8s TLS secret used for e2e test with expiry set to 365 days.
The test certs under deploy/example/csi-driver/testdata/certs are expired on Jan 29 2026

failing gh action https://github.com/kubernetes-csi/external-snapshot-metadata/actions/runs/21478246753/job/61868344625?pr=202

Error: GetMetadataAllocated(default,snap-1): rpc error: code = Unavailable desc = connection error: desc = "transport: authentication handshake failed: tls: failed to verify certificate: x509: certificate has expired or is not yet valid: current time 2026-01-29T12:43:24Z is after 2026-01-28T14:39:02Z"

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

@k8s-ci-robot k8s-ci-robot added release-note-none Denotes a PR that doesn't merit a release note. do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. kind/failing-test Categorizes issue or PR as related to a consistently or frequently failing test. labels Jan 29, 2026
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Jan 29, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @iPraveenParihar. Thanks for your PR.

I'm waiting for a kubernetes-csi member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot k8s-ci-robot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Jan 29, 2026
@iPraveenParihar
iPraveenParihar marked this pull request as ready for review January 29, 2026 14:58
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Jan 29, 2026
@Rakshith-R

Copy link
Copy Markdown
Contributor

/ok-to-test

@k8s-ci-robot k8s-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Jan 29, 2026
@iPraveenParihar
iPraveenParihar force-pushed the test/update-certs-and-tls-secret branch 2 times, most recently from 8b2cd02 to 00e5e1a Compare January 29, 2026 17:12

@PrasadG193 PrasadG193 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/LGTM
/approve

@k8s-ci-robot k8s-ci-robot added lgtm "Looks good to me", indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Jan 29, 2026
@iPraveenParihar
iPraveenParihar force-pushed the test/update-certs-and-tls-secret branch from 00e5e1a to dc5d6d9 Compare January 29, 2026 17:41
@k8s-ci-robot k8s-ci-robot removed the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 29, 2026
This commit updates certs and k8s tls secret used for e2e test
with expiry set to 365 days.
Earlier certs had validity of 60 days which were expired.

Signed-off-by: Praveen M <m.praveen@ibm.com>
@iPraveenParihar
iPraveenParihar force-pushed the test/update-certs-and-tls-secret branch from dc5d6d9 to fff6533 Compare January 30, 2026 08:03

@Rakshith-R Rakshith-R left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 30, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: iPraveenParihar, PrasadG193, Rakshith-R

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [PrasadG193,Rakshith-R]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit 719d8ba into kubernetes-csi:main Jan 30, 2026
9 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/failing-test Categorizes issue or PR as related to a consistently or frequently failing test. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesn't merit a release note. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants