Skip to content

feat(fuselage): Add missing message body additional styles to MessageBody - #1221

Merged
dougfabris merged 5 commits into
RocketChat:mainfrom
theivess:feat/blockquote-styles
Jan 10, 2024
Merged

feat(fuselage): Add missing message body additional styles to MessageBody#1221
dougfabris merged 5 commits into
RocketChat:mainfrom
theivess:feat/blockquote-styles

Conversation

@theivess

@theivess theivess commented Nov 12, 2023

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

This pull request addresses a TODO task related to code organization and style improvements
I have moved the relevant style to the designated Fuselage MessageBody component, resolving the outstanding task.


Screenshot 2023-11-13 at 2 40 42 AM

Issue(s)

TODO Link: Link to TODO

Further comments

None

@changeset-bot

changeset-bot Bot commented Nov 12, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 31f0a88

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@rocket.chat/fuselage Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@CLAassistant

CLAassistant commented Nov 12, 2023

Copy link
Copy Markdown

CLA assistant check
All committers have signed the CLA.

@tassoevan
tassoevan requested a review from a team November 16, 2023 18:11
dougfabris
dougfabris previously approved these changes Jan 10, 2024
@dougfabris dougfabris changed the title feat(fuselage): migrating blockquote styles to fuselage MessageBody component feat(fuselage): Migrate blockquote styles to fuselage MessageBody Jan 10, 2024
@dougfabris dougfabris changed the title feat(fuselage): Migrate blockquote styles to fuselage MessageBody feat(fuselage): Add missing message body additional styles to MessageBody Jan 10, 2024
dougfabris
dougfabris previously approved these changes Jan 10, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants