Skip to content

fix: call button misaligned in the contact information contextual bar - #33914

Closed
aleksandernsilva wants to merge 2 commits into
developfrom
fix/contact-info-call-button
Closed

fix: call button misaligned in the contact information contextual bar#33914
aleksandernsilva wants to merge 2 commits into
developfrom
fix/contact-info-call-button

Conversation

@aleksandernsilva

Copy link
Copy Markdown
Contributor

Proposed changes (including videos or screenshots)

This PR fixes the call button being misaligned in the contact information contextual bar. The issue seem to have been introduced with the ButtonGroup normalization #31499

Before:

Screenshot 2024-11-07 at 17 40 16

After:

Screenshot 2024-11-07 at 17 43 18

Issue(s)

Steps to test or reproduce

  • Acess workspace
  • Enabled Omnichannel VoIP in Workspace > Settings > Omnichannel voice channel (VoIP)
  • Open a livechat room
  • Open contact information contextual bar
  • Call button should be visible

Further comments

@dionisio-bot

dionisio-bot Bot commented Nov 7, 2024

Copy link
Copy Markdown
Contributor

Looks like this PR is not ready to merge, because of the following issues:

  • This PR is missing the 'stat: QA assured' label
  • This PR is missing the required milestone or project

Please fix the issues and try again

If you have any trouble, please check the PR guidelines

@changeset-bot

changeset-bot Bot commented Nov 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: eaabbab

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 35 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/api-client Patch
@rocket.chat/apps Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/ddp-client Patch
@rocket.chat/freeswitch Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/network-broker Patch
@rocket.chat/models Patch
@rocket.chat/ui-avatar Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/ui-voip Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Nov 7, 2024

Copy link
Copy Markdown
Contributor
PR Preview Action v1.4.8
🚀 Deployed preview to https://RocketChat.github.io/Rocket.Chat/pr-preview/pr-33914/
on branch gh-pages at 2024-11-07 20:56 UTC

@codecov

codecov Bot commented Nov 7, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 75.20%. Comparing base (322bafd) to head (eaabbab).
Report is 403 commits behind head on develop.

Additional details and impacted files

Impacted file tree graph

@@           Coverage Diff            @@
##           develop   #33914   +/-   ##
========================================
  Coverage    75.20%   75.20%           
========================================
  Files          495      495           
  Lines        21595    21595           
  Branches      5362     5362           
========================================
  Hits         16240    16240           
  Misses        4713     4713           
  Partials       642      642           
Flag Coverage Δ
unit 75.20% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@aleksandernsilva
aleksandernsilva marked this pull request as ready for review November 7, 2024 22:22
@aleksandernsilva
aleksandernsilva requested a review from a team as a code owner November 7, 2024 22:22
@aleksandernsilva aleksandernsilva added this to the 7.1.0 milestone Nov 7, 2024
@dougfabris

Copy link
Copy Markdown
Member

@aleksandernsilva I don't think its relevant since we're about to merge SCI which is going to change this entire contact view

@aleksandernsilva
aleksandernsilva marked this pull request as draft November 8, 2024 17:53
@aleksandernsilva aleksandernsilva removed this from the 7.1.0 milestone Nov 12, 2024
@aleksandernsilva
aleksandernsilva deleted the fix/contact-info-call-button branch February 13, 2025 15:29
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants