Skip to content

fix(livechat): registering guest multiple times cause message loss - #32300

Merged
kodiakhq[bot] merged 1 commit into
release-6.6.7from
backport-6.6.7-32069
Apr 23, 2024
Merged

fix(livechat): registering guest multiple times cause message loss#32300
kodiakhq[bot] merged 1 commit into
release-6.6.7from
backport-6.6.7-32069

Conversation

@dionisio-bot

@dionisio-bot dionisio-bot Bot commented Apr 23, 2024

Copy link
Copy Markdown
Contributor

Backport of #32069

@dionisio-bot
dionisio-bot Bot requested review from a team as code owners April 23, 2024 14:47
@dionisio-bot
dionisio-bot Bot requested a review from ggazzo April 23, 2024 14:47
@changeset-bot

changeset-bot Bot commented Apr 23, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2efcefc

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 30 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/livechat Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dionisio-bot

dionisio-bot Bot commented Apr 23, 2024

Copy link
Copy Markdown
Contributor Author

Looks like this PR is ready to merge! 🎉
If you have any trouble, please check the PR guidelines

@codecov

codecov Bot commented Apr 23, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.54%. Comparing base (27b5e32) to head (2efcefc).

Additional details and impacted files

Impacted file tree graph

@@              Coverage Diff               @@
##           release-6.6.7   #32300   +/-   ##
==============================================
  Coverage          54.54%   54.54%           
==============================================
  Files               2282     2282           
  Lines              50282    50282           
  Branches           10273    10273           
==============================================
  Hits               27424    27424           
- Misses             20366    20396   +30     
+ Partials            2492     2462   -30     
Flag Coverage Δ
unit 76.32% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@ggazzo
ggazzo force-pushed the backport-6.6.7-32069 branch 2 times, most recently from c4ac677 to 0a793b2 Compare April 23, 2024 18:27
@ggazzo
ggazzo force-pushed the backport-6.6.7-32069 branch from 0a793b2 to 2efcefc Compare April 23, 2024 18:32
@MartinSchoeler MartinSchoeler added the stat: QA assured Means it has been tested and approved by a company insider label Apr 23, 2024
@ggazzo ggazzo added this to the 6.6 milestone Apr 23, 2024
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Apr 23, 2024
@kodiakhq
kodiakhq Bot merged commit dcf4349 into release-6.6.7 Apr 23, 2024
@kodiakhq
kodiakhq Bot deleted the backport-6.6.7-32069 branch April 23, 2024 18:55
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: QA assured Means it has been tested and approved by a company insider stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants