Skip to content

fix: Always show all rooms when requesting chat history, even unserved ones - #31904

Merged
kodiakhq[bot] merged 8 commits into
developfrom
fix/chat-history-for-all-closed-rooms
Mar 18, 2024
Merged

fix: Always show all rooms when requesting chat history, even unserved ones#31904
kodiakhq[bot] merged 8 commits into
developfrom
fix/chat-history-for-all-closed-rooms

Conversation

@KevLehman

@KevLehman KevLehman commented Mar 5, 2024

Copy link
Copy Markdown
Member

https://rocketchat.atlassian.net/browse/SUP-394

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Mar 5, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5f656e5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 31 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/apps Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/model-typings Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/omnichannel-services Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/presence Patch
rocketchat-services Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/models Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch
@rocket.chat/instance-status Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Mar 5, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.60%. Comparing base (5bcf8ec) to head (5f656e5).

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31904      +/-   ##
===========================================
+ Coverage    54.59%   54.60%   +0.01%     
===========================================
  Files         2295     2294       -1     
  Lines        50575    50544      -31     
  Branches     10327    10321       -6     
===========================================
- Hits         27612    27601      -11     
+ Misses       20466    20445      -21     
- Partials      2497     2498       +1     
Flag Coverage Δ
e2e 53.68% <ø> (-0.02%) ⬇️
e2e-api 40.15% <ø> (+0.06%) ⬆️
unit 75.45% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@KevLehman
KevLehman marked this pull request as ready for review March 6, 2024 17:51
@KevLehman
KevLehman requested review from a team as code owners March 6, 2024 17:51
@dougfabris dougfabris changed the title fix: Show always all rooms when requesting chat history, even unserved ones fix: Always show all rooms when requesting chat history, even unserved ones Mar 6, 2024
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts
@scuciatto scuciatto added this to the 7.0 milestone Mar 9, 2024
Comment thread apps/meteor/server/models/raw/LivechatRooms.ts Outdated
Comment thread apps/meteor/server/models/raw/LivechatRooms.ts
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts Outdated
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts
KevLehman and others added 2 commits March 12, 2024 11:05
Co-authored-by: Marcos Spessatto Defendi <marcos.defendi@rocket.chat>
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts Outdated
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Mar 14, 2024
@scuciatto scuciatto removed this from the 7.0 milestone Mar 14, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 18, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 18, 2024
@kodiakhq
kodiakhq Bot merged commit 13fbf57 into develop Mar 18, 2024
@kodiakhq
kodiakhq Bot deleted the fix/chat-history-for-all-closed-rooms branch March 18, 2024 22:07
@scuciatto scuciatto added this to the 6.7 milestone Mar 19, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: ready to merge PR tested and approved waiting for merge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants