Skip to content

fix: livechat/visitor not updating custom fields some times - #31890

Merged
kodiakhq[bot] merged 12 commits into
developfrom
fix/customfields-sometimes-not-processing
Mar 22, 2024
Merged

fix: livechat/visitor not updating custom fields some times#31890
kodiakhq[bot] merged 12 commits into
developfrom
fix/customfields-sometimes-not-processing

Conversation

@KevLehman

@KevLehman KevLehman commented Mar 4, 2024

Copy link
Copy Markdown
Member

https://rocketchat.atlassian.net/browse/SUP-488

Proposed changes (including videos or screenshots)

Issue(s)

Steps to test or reproduce

Further comments

@changeset-bot

changeset-bot Bot commented Mar 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 52d7564

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 31 packages
Name Type
@rocket.chat/meteor Patch
@rocket.chat/model-typings Patch
@rocket.chat/apps Patch
@rocket.chat/models Patch
@rocket.chat/account-service Patch
@rocket.chat/authorization-service Patch
@rocket.chat/ddp-streamer Patch
@rocket.chat/omnichannel-transcript Patch
@rocket.chat/presence-service Patch
@rocket.chat/queue-worker Patch
@rocket.chat/stream-hub-service Patch
@rocket.chat/omnichannel-services Patch
rocketchat-services Patch
@rocket.chat/core-services Patch
@rocket.chat/cron Patch
@rocket.chat/instance-status Patch
@rocket.chat/presence Patch
@rocket.chat/core-typings Patch
@rocket.chat/rest-typings Patch
@rocket.chat/gazzodown Patch
@rocket.chat/livechat Patch
@rocket.chat/ui-contexts Patch
@rocket.chat/api-client Patch
@rocket.chat/license Patch
@rocket.chat/pdf-worker Patch
@rocket.chat/ddp-client Patch
@rocket.chat/fuselage-ui-kit Patch
@rocket.chat/ui-client Patch
@rocket.chat/ui-video-conf Patch
@rocket.chat/uikit-playground Patch
@rocket.chat/web-ui-registration Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@codecov

codecov Bot commented Mar 4, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 54.34%. Comparing base (469afbd) to head (52d7564).

Additional details and impacted files

Impacted file tree graph

@@             Coverage Diff             @@
##           develop   #31890      +/-   ##
===========================================
- Coverage    54.70%   54.34%   -0.36%     
===========================================
  Files         2298     2247      -51     
  Lines        50702    49799     -903     
  Branches     10352    10111     -241     
===========================================
- Hits         27734    27063     -671     
+ Misses       20472    20280     -192     
+ Partials      2496     2456      -40     
Flag Coverage Δ
e2e 52.97% <ø> (-0.73%) ⬇️
e2e-api 40.16% <ø> (+0.06%) ⬆️
unit 75.74% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@KevLehman
KevLehman marked this pull request as ready for review March 6, 2024 18:05
@KevLehman
KevLehman requested review from a team as code owners March 6, 2024 18:05
Comment thread apps/meteor/app/livechat/server/api/v1/visitor.ts
Comment thread apps/meteor/app/livechat/server/api/v1/visitor.ts
Comment thread apps/meteor/app/livechat/server/api/v1/visitor.ts Outdated
@scuciatto scuciatto added this to the 7.0 milestone Mar 9, 2024
Comment thread apps/meteor/app/livechat/server/api/v1/visitor.ts
Comment thread apps/meteor/app/livechat/server/api/v1/visitor.ts
Comment thread apps/meteor/app/livechat/server/api/v1/visitor.ts
Comment thread apps/meteor/app/livechat/server/api/v1/visitor.ts
…m:RocketChat/Rocket.Chat into fix/customfields-sometimes-not-processing
Comment thread apps/meteor/tests/end-to-end/api/livechat/09-visitors.ts
@scuciatto scuciatto removed this from the 7.0 milestone Mar 14, 2024
@dionisio-bot dionisio-bot Bot added the stat: ready to merge PR tested and approved waiting for merge label Mar 21, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 21, 2024
@dionisio-bot dionisio-bot Bot added stat: ready to merge PR tested and approved waiting for merge and removed stat: ready to merge PR tested and approved waiting for merge labels Mar 21, 2024
@kodiakhq
kodiakhq Bot merged commit 0570f67 into develop Mar 22, 2024
@kodiakhq
kodiakhq Bot deleted the fix/customfields-sometimes-not-processing branch March 22, 2024 12:51
@scuciatto scuciatto added this to the 6.8 milestone Apr 5, 2024
@dionisio-bot

dionisio-bot Bot commented Apr 5, 2024

Copy link
Copy Markdown
Contributor

Looks like this PR is not ready to merge, because of the following issues:

  • This PR is not mergeable
    Please fix the issues and try again
    If you have any trouble, please check the PR guidelines

@dionisio-bot dionisio-bot Bot added stat: QA assured Means it has been tested and approved by a company insider and removed stat: ready to merge PR tested and approved waiting for merge stat: QA tested labels Apr 5, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

stat: QA assured Means it has been tested and approved by a company insider

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants