feat(livechat): hide or change logo - #31820
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: c20ddbf The changes in this PR will be included in the next version bump. This PR includes changesets to release 32 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov ReportAll modified and coverable lines are covered by tests ✅
Additional details and impacted files@@ Coverage Diff @@
## develop #31820 +/- ##
===========================================
+ Coverage 54.70% 54.74% +0.03%
===========================================
Files 2291 2299 +8
Lines 50618 50733 +115
Branches 10337 10360 +23
===========================================
+ Hits 27690 27773 +83
- Misses 20443 20465 +22
- Partials 2485 2495 +10
Flags with carried forward coverage won't be shown. Click here to find out more. |
aleksandernsilva
force-pushed
the
feat/livechat-hide-change-logo
branch
2 times, most recently
from
February 27, 2024 16:15
b201ec2 to
37a209f
Compare
aleksandernsilva
marked this pull request as ready for review
February 27, 2024 18:46
MarcosSpessatto
suggested changes
Feb 28, 2024
ggazzo
requested changes
Feb 29, 2024
ggazzo
left a comment
Member
There was a problem hiding this comment.
instead of creating a tag property and spreading it across all components (as well as the respective margins) why not pass it through the label property?
ggazzo
requested changes
Feb 29, 2024
aleksandernsilva
marked this pull request as draft
February 29, 2024 14:16
aleksandernsilva
force-pushed
the
feat/livechat-hide-change-logo
branch
from
February 29, 2024 14:29
e6ed8ff to
e9b41f8
Compare
aleksandernsilva
marked this pull request as ready for review
March 1, 2024 14:19
KevLehman
requested changes
Mar 4, 2024
aleksandernsilva
force-pushed
the
feat/livechat-hide-change-logo
branch
from
March 7, 2024 19:40
afaaa08 to
ed7001b
Compare
aleksandernsilva
marked this pull request as draft
March 7, 2024 20:41
aleksandernsilva
marked this pull request as ready for review
March 7, 2024 22:26
aleksandernsilva
marked this pull request as draft
March 13, 2024 12:58
aleksandernsilva
force-pushed
the
feat/livechat-hide-change-logo
branch
2 times, most recently
from
March 13, 2024 13:04
0f10639 to
7f92be1
Compare
aleksandernsilva
dismissed stale reviews from MarcosSpessatto, tassoevan, and KevLehman
via
March 22, 2024 20:46
0636587
aleksandernsilva
force-pushed
the
feat/livechat-hide-change-logo
branch
from
March 22, 2024 20:46
1ec991c to
0636587
Compare
aleksandernsilva
requested review from
KevLehman,
MarcosSpessatto and
tassoevan
March 22, 2024 20:51
tassoevan
previously approved these changes
Mar 22, 2024
KevLehman
previously approved these changes
Mar 22, 2024
ggazzo
reviewed
Mar 22, 2024
ggazzo
force-pushed
the
feat/livechat-hide-change-logo
branch
from
March 22, 2024 21:54
80d6c61 to
a3a807e
Compare
ggazzo
force-pushed
the
feat/livechat-hide-change-logo
branch
from
March 22, 2024 23:26
a3a807e to
c20ddbf
Compare
MarcosSpessatto
approved these changes
Mar 25, 2024
ggazzo
approved these changes
Mar 25, 2024
Merged
Salvatore79999
approved these changes
Apr 10, 2024
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Proposed changes (including videos or screenshots)
This PR adds two new Livechat settings:
ℹ️ Tests for both settings will be implemented in separate tasks : PROSVC-80 and PROSVC-81
Issue(s)
PROSVC-48
Steps to test or reproduce
Further comments